)]}'
{"id":"openstack%2Fautomaton~751431","triplet_id":"openstack%2Fautomaton~master~Iddd80d10bad24bf52c72401fb49fddcf8335060b","project":"openstack/automaton","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"Iddd80d10bad24bf52c72401fb49fddcf8335060b","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-11 20:54:38.000000000","updated":"2020-09-14 16:36:36.000000000","submitted":"2020-09-14 16:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751431-1600100668533-c99f38e4","meta_rev_id":"bc0bf84991000cc9f64f3a6c5408d698a78eb2ef","_number":751431,"virtual_id_number":751431,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 16:24:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 20:55:29.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-14 08:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-14 12:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 20:55:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-14 12:14:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-14 16:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46de5fa0f75526723aa7f322ec095ce1be5b4dd6","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-11 20:54:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c44cab97038cc9e3d355d40fd5525711485c6ab6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 20:55:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9eb011ac9a5026e8b5ce88a45f2499b6fa26d830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 22:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db9c986db8984a2689a726041cdb8d42 : SUCCESS in 1h 04m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a8dbb70361e4b559ad79417c5f8c04e : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d21b87cdd64dc09632e5a1ef9e922d : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/364606735fd54e7481e57c087757b8ac : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de258cda9834412f9512e29d2dbd5774 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9739a414e74468a3cf14784eadb228 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"559db5daaa8159034ae4c6f8b9720f58e31b92db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 08:27:36.000000000","message":"Patch Set 1: Code-Review+2\n\nAs we are in FF waiting a bit before merging these changes to avoid conflicts","accounts_in_message":[],"_revision_number":1},{"id":"e9bc94acfda9ff1d39eb1b233742410edcfed63a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 12:14:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e287964bda489cad9d1e3ce8ec3687ad9a0c5a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:43:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2bcebac69534b4a01ab12a009d3ef8deee2110a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 16:24:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c34097a672ed71d4dbecc280acf73962fc012241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 16:24:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7ee597d78e4419b804607e5a84541db : SUCCESS in 1h 15m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/114f5ad77cc147f2b320a72acf3a46ea : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4fb7ce35484c108824b50f00a747ca : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c87573117a47118ee00b22019e331c : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d6533dd22641f5bf461395d72c37f9 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66d3d9e6c644658a92e6edbf97fc540 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ef7ab14dc68526a9eb4032a3afa898e1c8bd6d06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 16:36:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9968e73f5f614eaa9edf5441193d844f : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74bef8149e7beb17929c316fa18a6a226a3a875","revisions":{"f74bef8149e7beb17929c316fa18a6a226a3a875":{"kind":"REWORK","_number":1,"created":"2020-09-11 20:54:38.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/31/751431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/31/751431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/31/751431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/31/751431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/31/751431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/31/751431/1"}}},"commit":{"parents":[{"commit":"5f57b8543f61e1004e8852576774eb10b2bb555f","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/5f57b8543f61e1004e8852576774eb10b2bb555f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 20:54:36.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 20:54:36.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Iddd80d10bad24bf52c72401fb49fddcf8335060b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/f74bef8149e7beb17929c316fa18a6a226a3a875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/f74bef8149e7beb17929c316fa18a6a226a3a875"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
