)]}'
{"id":"openstack%2Fautomaton~753236","triplet_id":"openstack%2Fautomaton~master~Iaa239cde7d159f3098fdc22c826ee4dc4c52332a","project":"openstack/automaton","branch":"master","hashtags":[],"change_id":"Iaa239cde7d159f3098fdc22c826ee4dc4c52332a","subject":"Bump hacking min version to 3.0.1","status":"ABANDONED","created":"2020-09-22 07:12:12.000000000","updated":"2020-12-18 09:01:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c2542f7fd1f47b4d8af4df161111bc63383edd9","_number":753236,"virtual_id_number":753236,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-16 01:02:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 08:16:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-22 09:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 14:30:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"7fea3b3a3ee751a09b2c10c25ce108ecc0b4a820","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-22 07:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a92e4e007c6ddb1b5b16a791d7309a2a0b7f7bbf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-22 08:16:14.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"3562299bee9c2a7798d25b1c575ab36a295200f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 08:22:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/77207e7d70a8425fb54246d7249cb859 : SUCCESS in 5m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f210075a32574bafaf075f7b55d6dd35 : FAILURE in 1h 02m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1591f6cc08c1486cb50d41790da7e449 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2402978d9dd49bebf82741e8ec67f26 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/110dfdd7a5ff4503b3d745f0f4e19fb9 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc9b6f4d5c0c4d9fb6f771dbbb08b338 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1acc38953c435eb9114ba38847e92a : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6a7e3a481e498e86cf901ae9467b3b : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"63dc4f02ea044bc22bc29f2644ab0b90eb0ce8bd","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-22 08:23:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f770f568d03c478bc71481a83b137edf0a4427bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 09:41:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8dc082b9c5be47feb5d0a548c126b7d3 : SUCCESS in 3m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fda244384c6c42fa97d966508ba61303 : SUCCESS in 1h 11m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca7f240ccb0044f0b702cdd9f22d349c : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a404c29ecd28439198d4ddf39fdaaad8 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d69a6a327b54fb4bb6626cc0e7bd60e : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a42653e1f6443187e8e67b4afd67e7 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71502bd89488490eb58e148d3631738d : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a4e2cff0474d088737eedc8314df4b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5c8b036d4d8294877cc72733b6486c7b6fbaa42b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 14:30:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3488dca59efb3f418c1d8be8179a6487e8df7a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 14:31:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"033186c42eac7cbde9accb030cc29bc674c6140f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 16:24:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71c403a693f14a6d89b5a61ea21009e2 : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cad7753c1eb480a835bf075fb645250 : FAILURE in 1h 46m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/687fa69ba35a425c9bbb402efc6bafd4 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6fc55fe0dc4291ad9a8f205e33eb20 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4b1d87628d4818a21cc1d71c686992 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45ca0c8613f4aae94850e4602c38558 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3917d4e8a02b4ab9a6d433f52e8d25fe : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa665c708d6744a0982fad08998baf80 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ab8eaa06807890b49076581b3978b0e3e5e806be","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-14 00:58:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ccc304ea33457a4a1c121ea0f841a7f0a937efed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 02:03:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e1dd34e12b4409197f7e5d869aead2a : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/894ab659488e4b83bccd0778d415ab9c : SUCCESS in 1h 03m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a088a4fd4f2c4baba4ee0f17485b4256 : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d99d03bd8e471cb0f127bcf6d7e89b : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e290e42e52f54d03a4e89c3a70459254 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bbfa1db34c143a083611752b3dcef25 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/292219ac83994f19b998e3b3474a7f92 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6df85f6afb24bb89ae9ece1caaad3e1 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8262c787c0c24bd9a4c39154f1373c04 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b91c7671d81094c7f6a6aec306aea861f5cfda12","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-15 01:27:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5256608a2d6930345c46eba1637904beeeaad886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 02:36:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f05479d27074556bf989cc555ff132a : SUCCESS in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9c02fcecb2746d4893c6078eacf1acb : SUCCESS in 1h 07m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1405f30f91b48fdae90e3b9d89bce8a : FAILURE in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6cb5c24646a45a09e122614725c7fa6 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90bcc6e0459495e872a0d375a9ba809 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c567d11662664804acd05418b19db195 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4b596593304a4f96eb9e45683e1428 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e1dc05fdc34972b7cd430124d03b40 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea2a7c62cf14947839a9a1a58aedc0a : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bc995d653b9fce60ba937743ebf3a048929e22a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-15 05:54:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a005fd0f28cb898474a3528a9550a60bde2fbd79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 07:39:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2dbacb1460154f35a9aea040d037f71d : SUCCESS in 4m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37cda482d3fe48d0b830010b079f622a : SUCCESS in 1h 39m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19fd69d5ba1c41beb1b2bb1e88d36638 : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f0f37394c14a349ab3c97eb7f7b4ba : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e928e86157424e709e63b575b28fef3d : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdfa38901e374f8da8bf0030d070ecf9 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6800e2526634aca82ab80ab3e7f77cb : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372c120f389248afa07679b5a008cf1e : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2be15df892942c0b9bc42d5f5d82668 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"35896f70e7448ba8bd1aa425bc8c8e87feb584eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-15 08:17:10.000000000","message":"Patch Set 2:\n\nLower constraints should be fixed first","accounts_in_message":[],"_revision_number":2},{"id":"3807ee70a42436f240aafedf77bf37b7995b67e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-16 01:02:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"777a149a5aff693e0c4370e0ae440376ab3f578a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 01:02:42.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"bebc6db6e2e92a5acc924b56333edec1029e50c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-16 09:55:52.000000000","message":"Patch Set 3:\n\nHello,\n\nYou\u0027ve few issues here:\n- You\u0027ve a merge conflict due to https://review.opendev.org/c/openstack/automaton/+/766221 that removed hacking from test-requirements.txt\n- Our lower-constraints job is broken due to pip\u0027s new resolver (c.f http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019149.html) and this should be addressed first.\n\nI would really prefer to address the pip resolver issue through a separated patch to avoid to mix topics, however until this problem isn\u0027t solved we aren\u0027t able to merge your patches as our gate is broken.\n\nHopefully it will help you.","accounts_in_message":[],"_revision_number":3},{"id":"3c2542f7fd1f47b4d8af4df161111bc63383edd9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-18 09:01:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0a7a31257ed3d31e615c384c62e3091fc7f4b877","revisions":{"e3c277612ab7687fd7d97bdf911ae75aba18d784":{"kind":"REWORK","_number":1,"created":"2020-09-22 07:12:12.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/36/753236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/36/753236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/36/753236/1"}}},"commit":{"parents":[{"commit":"f74bef8149e7beb17929c316fa18a6a226a3a875","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/f74bef8149e7beb17929c316fa18a6a226a3a875"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-22 07:11:21.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-22 07:12:08.000000000","tz":480},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: Iaa239cde7d159f3098fdc22c826ee4dc4c52332a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/e3c277612ab7687fd7d97bdf911ae75aba18d784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/e3c277612ab7687fd7d97bdf911ae75aba18d784"}]},"branch":"refs/heads/master"},"7afaf84800d311099e05e7ea5b4adccb9f8b16ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-15 05:54:41.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/36/753236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/36/753236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/36/753236/2"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-22 07:11:21.000000000","tz":480},"committer":{"name":"wushiming","email":"wushiming@yovole.com","date":"2020-12-15 05:54:41.000000000","tz":0},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: Iaa239cde7d159f3098fdc22c826ee4dc4c52332a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/7afaf84800d311099e05e7ea5b4adccb9f8b16ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/7afaf84800d311099e05e7ea5b4adccb9f8b16ac"}]},"branch":"refs/heads/master","description":"Rebase"},"0a7a31257ed3d31e615c384c62e3091fc7f4b877":{"kind":"REWORK","_number":3,"created":"2020-12-16 01:02:28.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/36/753236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/36/753236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/36/753236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/36/753236/3"}}},"commit":{"parents":[{"commit":"f74bef8149e7beb17929c316fa18a6a226a3a875","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/f74bef8149e7beb17929c316fa18a6a226a3a875"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-22 07:11:21.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-16 01:02:17.000000000","tz":480},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: Iaa239cde7d159f3098fdc22c826ee4dc4c52332a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/0a7a31257ed3d31e615c384c62e3091fc7f4b877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/0a7a31257ed3d31e615c384c62e3091fc7f4b877"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
