)]}'
{"id":"openstack%2Fautomaton~766221","triplet_id":"openstack%2Fautomaton~master~Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a","project":"openstack/automaton","branch":"master","topic":"remove-six","hashtags":[],"change_id":"Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a","subject":"Remove six dependency","status":"MERGED","created":"2020-12-09 13:51:16.000000000","updated":"2020-12-15 14:33:35.000000000","submitted":"2020-12-15 14:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766221-remove-six","meta_rev_id":"627b0794189b9d28bf555703b9101f4cd839eb74","_number":766221,"virtual_id_number":766221,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 14:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-12-15 10:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-12-14 10:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-12-15 10:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 13:55:27.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2020-12-09 15:29:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-12-09 16:08:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 14:14:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-12-14 10:44:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"5540fc0c408e5054eaff02362b616de389ef409c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-09 13:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db5a06029196f95f55e0b201d7eae7690fb4610","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-09 13:55:27.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2cfd34af0f8dcc52d0c4eb12250cfc5a8bc1aa10","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-09 14:42:26.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ec519facf361669ae08aa545dadcfd131a2d39d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-09 14:42:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc3613cb3a409ff15ef5d74b84ce5d7926e3c2c0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-09 15:29:50.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"db19df8c5422b550b11ae91cd3c3d170c156c717","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-09 15:29:58.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da5cf6a4bab8c8234ae1666cf1ae5fae3b5c500b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 16:08:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/88c72c3a5ef34ac2b11e54bda34a3a38 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96e38c47ab3343d09de258ba499775f1 : SUCCESS in 1h 16m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea09b6c83aa74c829ffda8c74de10c13 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdba3da15ada4a48bb1736be7e7b1e8e : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4161a02bd9c4476b04eab9701b62971 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dcedcc36c73483bb8e6ca2c318903a0 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/768f0d8f89f24776952b6714d51aa8cc : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f11ef0e07ba4cf4948c2403689ad451 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"82e836fcbce84717e619c8c5b697a5c7185b423e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-09 16:30:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59e896766babf7bcb29be6e5b593554cedec4124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 18:38:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81952243228f42ba82bc907f8fd88a59 : SUCCESS in 2m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe2b423a16f94cad892b42454e4054a2 : SUCCESS in 1h 20m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f11f40d2826f4ca2abbff40359fda2bb : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed65287901e74c0c942ea7418d6040e5 : FAILURE in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47271ac36d604f029557f4be895eca06 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/427e5364b85f4d8b857529a12584662f : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d62a00cb0f854a0284062f548c13a68f : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963dafe6bd9749109adea3a90e634f3a : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664f2039167a4dce8946eddf6388b8f5 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"0d2317ab41505970358a74b72d4f879df44925a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-10 09:56:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"626ab429e1a65acf3e9fa465a92fe0b22dd47eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 11:18:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71dd354d2e2449769a3bb65356778562 : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad8a9b2af9bd4f589f4a77963423cfa0 : SUCCESS in 1h 19m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b104d30f62534a298edfc5e05f0d4c37 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060ff1b092b94b8fa3c1f5100d70e6db : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88465e01d03a4a3db9a2e8cc1484ed5b : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74bf9343ff5b4d85b79791d5bcff37ac : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a450581d14a4fb9926abf53dd2e82aa : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2554a928bda24dd0b58976d28c58c673 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66583c7ac7cb493c880b253014e0ede5 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":4},{"id":"8e57de73b0555cc6be27a11acc4bdc6cfa93402b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 14:14:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\n-1 is for the tag discrepancy. Everything else looks good","accounts_in_message":[],"_revision_number":4},{"id":"37bbfe6b740549907dbfa25740fd4ef2b2145e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-12 07:08:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"be70a097e4a11a0757586dc75cfbc400e908e880","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-12 07:08:43.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0c2808ac53cf5024ae36033c0f6c52228d14e25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 08:42:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/38de7ecf8f044c2d8dce10d0bddd731c : SUCCESS in 2m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48884d0f25dd4f5dbd15c8382ba63f26 : SUCCESS in 1h 28m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fd415bc66d5422a92afabe1816149f9 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0580f140c4401b88e0ae83b3d0baae : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53134fa16c940319e262dca5e928084 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21d198109af84d538f4639f0b9e3cc31 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d9c76f7d894658af0c20ef34e7f3e9 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3cf698d9934da0baca969397507b41 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fff6aaf24b24118998b8ef16d3ada61 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"1c9827f9ac0ead44c03b356a7dea108163b49d24","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-14 10:44:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"060795447f81c79722df9a6bc4d50e4b38f08811","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-15 10:04:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3fbbe18d0bd92f7a8444f42eb923cf7c3296e29b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-15 10:04:19.000000000","message":"Patch Set 5:\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"e30392dc2a37b305e04a7489e01b6a47965a6afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 10:04:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cab586c5a49f0556ef5148ddaeb6cfaf91518b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 11:31:41.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dda38cb82f4041cbad36181a13ae7b5c : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b80931c7f7f94527a395c266bb31171b : FAILURE in 1h 24m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e72921132b7410f9875b306892e6e64 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf9378e82bb45499a6ef5fd986c9dca : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424e4576bb9842a69a666f2d85d980ae : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa6bd341ad8f4197a9a9d55456f34545 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663178eae27b401f84f89cfea3d6fc0c : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5eab2bf0b654e75b438ce000bbd2da0 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4cd9c62d0f02389bdc604aefc93329f640f0bdc4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-15 11:34:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"114a84e10b411d1e11524454bc511545b3e4e381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:01:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8da1130a99ea4a328494d187cb12547c : SUCCESS in 3m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00ea599984564ec5ba204cfa5d165414 : SUCCESS in 1h 12m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/768af55624094249841eed3eee178f8a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d74514abd9444ce8e6974906f1006da : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1351b0d514d42babf56a273a6be6ecc : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9837cf4ef73445af9f5bd65ff1ea688e : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/819542808f1d4d198d06f53a0c488104 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ad7f9dda9d44fb9cc16957a1cd04d6 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a826278816a94096828ebc386f44324b : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":5},{"id":"f068a0840471b66576f560c424055ec6aee09f81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:02:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"13e78fcb30d22c3ee30cf55c0f433b46e98b04d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:32:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2fc29febd934748a96a8af09867a8a5 : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9558ca0d955b4235985f5b34f8ff7983 : SUCCESS in 1h 23m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee55549c60f34535a49774ab66f55068 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da615a3f7f7944dc93909fc984125fe7 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7bd0cff3fa245c0801f14a9f28ad2e8 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/925a436309454946a42a2dc63e03c35a : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677921f9f38f4141b642b41d5d7fca7c : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a5f72b4a047436da857d0e55628b3f3 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":5},{"id":"24e625f9959be5c2e2968d94e7be3aa889b23cff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:32:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"627b0794189b9d28bf555703b9101f4cd839eb74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:33:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67d6a197db472bb7a253eb8852ea76 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e9ef802cf246059252fea17b9cd8e6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fed5f89755b7e787fa5a46ffa468784dd26f8213","revisions":{"aba79046312f67b591d685af4924d72de34c9874":{"kind":"REWORK","_number":1,"created":"2020-12-09 13:51:16.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/21/766221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/21/766221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/21/766221/1"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:50:43.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:51:14.000000000","tz":60},"subject":"Remove six dependency","message":"Remove six dependency\n\nWe run on Python 3 only, time to remove six dependency\n\nChange-Id: Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/aba79046312f67b591d685af4924d72de34c9874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/aba79046312f67b591d685af4924d72de34c9874"}]},"branch":"refs/heads/master"},"383427e91ca9de5c825e0d345c2f3c0c38b9d4e8":{"kind":"REWORK","_number":2,"created":"2020-12-09 14:42:34.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/21/766221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/21/766221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/21/766221/2"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:50:43.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 14:41:53.000000000","tz":60},"subject":"Remove six dependency","message":"Remove six dependency\n\nWe run on Python 3 only, time to remove six dependency\n\nChange-Id: Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/383427e91ca9de5c825e0d345c2f3c0c38b9d4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/383427e91ca9de5c825e0d345c2f3c0c38b9d4e8"}]},"branch":"refs/heads/master"},"1a38b4507817c0ed6c6ed4fe8ad34b3db77b41e3":{"kind":"REWORK","_number":3,"created":"2020-12-09 16:30:05.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/21/766221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/21/766221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/21/766221/3"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:50:43.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 16:29:28.000000000","tz":60},"subject":"Remove six dependency","message":"Remove six dependency\n\nWe run on Python 3 only, time to remove six dependency\n\nAlso fix l-c for new pip-resolver and move pep8 deps\nin tox.ini to avoid conflicts.\n\nChange-Id: Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/1a38b4507817c0ed6c6ed4fe8ad34b3db77b41e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/1a38b4507817c0ed6c6ed4fe8ad34b3db77b41e3"}]},"branch":"refs/heads/master"},"8696858ad5c7eb84ba06c556934582f0e375b0dc":{"kind":"REWORK","_number":4,"created":"2020-12-10 09:56:31.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/21/766221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/21/766221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/21/766221/4"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:50:43.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-10 09:55:57.000000000","tz":60},"subject":"Remove six dependency","message":"Remove six dependency\n\nWe run on Python 3 only, time to remove six dependency\n\nAlso fix l-c for new pip-resolver and move pep8 deps\nin tox.ini to avoid conflicts.\n\nAnd since we\u0027re here, update pre-commit config to support correct\nhacking version.\n\nChange-Id: Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/8696858ad5c7eb84ba06c556934582f0e375b0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/8696858ad5c7eb84ba06c556934582f0e375b0dc"}]},"branch":"refs/heads/master"},"fed5f89755b7e787fa5a46ffa468784dd26f8213":{"kind":"REWORK","_number":5,"created":"2020-12-12 07:08:14.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/21/766221/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/21/766221/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/21/766221/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/21/766221/5"}}},"commit":{"parents":[{"commit":"c6f5685abbab1a6f44827e72f8295c7cf6813fa0","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/c6f5685abbab1a6f44827e72f8295c7cf6813fa0"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-09 13:50:43.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-12 07:08:06.000000000","tz":60},"subject":"Remove six dependency","message":"Remove six dependency\n\nWe run on Python 3 only, time to remove six dependency\n\nAlso fix l-c for new pip-resolver and move pep8 deps\nin tox.ini to avoid conflicts.\n\nAnd since we\u0027re here, update pre-commit config to support correct\nhacking version.\n\nChange-Id: Ibbe733c4be37d1a9dd9149d491591ef2e50dad6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/fed5f89755b7e787fa5a46ffa468784dd26f8213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/fed5f89755b7e787fa5a46ffa468784dd26f8213"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
