)]}'
{"id":"openstack%2Fautomaton~774370","triplet_id":"openstack%2Fautomaton~master~I2a09920c988dd49943eda5e11a8a259d868d0169","project":"openstack/automaton","branch":"master","topic":"remove_version_limit","hashtags":[],"change_id":"I2a09920c988dd49943eda5e11a8a259d868d0169","subject":"Remove PrettyTable version limit","status":"ABANDONED","created":"2021-02-07 08:08:54.000000000","updated":"2021-02-18 02:46:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1b3e860aca26f4218ef8e842bc1402e1bf0ce1e","_number":774370,"virtual_id_number":774370,"owner":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-07 09:35:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 09:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 15:17:33.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"CC"}],"messages":[{"id":"cdb2aa6f58d9b5059f95e32119eb51d90299c1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-07 08:08:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2293d8f08c588722c235a96b46165791f8b39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:35:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe554dba7c894445ad73fa085844a41f : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/688d8941ad7c4c5cae046bd91ef82544 : SUCCESS in 1h 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd5aa019cd746d59d96ff4c074609b7 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b7f8f71aa646b2b3c2ecf637faeb68 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/597a8b413b674a7e82b05e2f7dad2db7 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f501ea4db08446084e971b301e8d3e3 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917fe2203368492ea36582f52463e5d1 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1dfbaeb1e4c74e19a8cc0987d1ffe5de04c721c2","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-02-08 15:17:33.000000000","message":"Patch Set 1:\n\nThanks. Don\u0027t we usually need to update requirements first though? https://opendev.org/openstack/requirements/src/commit/ed019e17dfc136092fe55ff41e2570f566e320e9/global-requirements.txt#L196","accounts_in_message":[],"_revision_number":1},{"id":"8fc63711d7b88b18346917b4c324a47babb3ecc8","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-18 01:56:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thanks. Don\u0027t we usually need to update requirements first though? https://opendev.org/openstack/requirements/src/commit/ed019e17dfc136092fe55ff41e2570f566e320e9/global-requirements.txt#L196\n\nYes, usually we should update openstack/requirements first. But it seems that we hit some CI upgrade issue there. https://review.opendev.org/c/openstack/requirements/+/774368\n\nStephen has a patch https://review.opendev.org/c/openstack/requirements/+/774917 But there is still some issue.\n\nThen Cliff bumped the version by hand: https://review.opendev.org/c/openstack/cliff/+/774915/1\n\nLet block this patch until the problem in  openstack/requirements is fixed. Or auomaton can do the same as Cliff","accounts_in_message":[],"_revision_number":1},{"id":"fd940946327f4939dd4ca6d6608a6d02cbc9404e","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-18 02:30:14.000000000","message":"Patch Set 1:\n\nOh, due to the CI issues in openstack/requirements, I think we should uncap the version limit in automaton first, like what Cliff did.","accounts_in_message":[],"_revision_number":1},{"id":"d1b3e860aca26f4218ef8e842bc1402e1bf0ce1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-18 02:46:42.000000000","message":"Abandoned\n\nJust find that automaton has uncap the limit already https://opendev.org/openstack/automaton/commit/406007e43eb48f959b25699423e9826b23ebe760","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3805b5603636d0a42a50afe73cd954957dcc1919","revisions":{"3805b5603636d0a42a50afe73cd954957dcc1919":{"kind":"REWORK","_number":1,"created":"2021-02-07 08:08:54.000000000","uploader":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"ref":"refs/changes/70/774370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/70/774370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/70/774370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/70/774370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/70/774370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/70/774370/1"}}},"commit":{"parents":[{"commit":"155c7c104e5f1c414b3562fe784f642f222aecae","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/155c7c104e5f1c414b3562fe784f642f222aecae"}]}],"author":{"name":"Wander Way","email":"wanderwayout@gmail.com","date":"2021-02-07 08:03:00.000000000","tz":0},"committer":{"name":"Wander Way","email":"wanderwayout@gmail.com","date":"2021-02-07 08:08:46.000000000","tz":0},"subject":"Remove PrettyTable version limit","message":"Remove PrettyTable version limit\n\nPrettyTable library is alive now. It has newer version already.\n\nSo it\u0027s good to keep the library newer.\n\nChange-Id: I2a09920c988dd49943eda5e11a8a259d868d0169\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3805b5603636d0a42a50afe73cd954957dcc1919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3805b5603636d0a42a50afe73cd954957dcc1919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
