)]}'
{"id":"openstack%2Fautomaton~782440","triplet_id":"openstack%2Fautomaton~master~I402e708832395d95b04a5b956a19d5cc8c3f1c1f","project":"openstack/automaton","branch":"master","topic":"move_flake8_local","hashtags":[],"change_id":"I402e708832395d95b04a5b956a19d5cc8c3f1c1f","subject":"Move flake8 as a pre-commit local target.","status":"MERGED","created":"2021-03-23 11:53:43.000000000","updated":"2021-04-16 14:01:00.000000000","submitted":"2021-04-16 13:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782440-move_flake8_local","meta_rev_id":"6049b087f7deaa833a6a8291315f59d70cf5b0bc","_number":782440,"virtual_id_number":782440,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 13:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-04-16 11:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2021-04-16 11:08:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 13:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 13:56:57.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-24 08:46:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-16 11:07:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ab0aa8cd6c241db73c1c7f60bd1f67ff1a7c8ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-23 11:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13d7e344dd1c299e792032a86ef87b9d5881d35d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:22:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f719f73fb1884eb3b3f45d2be2484bf2 : SUCCESS in 1h 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb47137f14d4f008f12a0080704e536 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff3927ae414047699777cb9db7a7349b : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94b93987e69c4467ad1c60b886fe1c99 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b3f9624313146c9b9ea654654a325d9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4e7c37dcf34beb96b70fba0fac5f10 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cbd58e478ad15c5ba9938d446f9a48d623da452c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-23 13:56:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37e6edf86cc32652be493dafe1ac2e025c182877","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 08:46:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed02be967b9dc69be11ce8409af3ab74a98e19c8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-07 10:16:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2d4fbf50080b2cfce21a491aa7943145f99515e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-07 14:39:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95e6ad322cac48f7b7993d21ea599f73608f552d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:45:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/959581dc6eda4ac583e439748a412c00 : SUCCESS in 2h 03m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f65dd7922b4c769a978095bb8e303a : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa64ed6295e04684a78c9091df6fafd7 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3d65fa693847569a68af163b1c69c6 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8db716ee26348e8b12275ed186ef729 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4912f8a0b45435cacb8ba80021e2d11 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b205a52e43f6b6ba81a39ad9bdcc496d52bfa2a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-15 13:02:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"027129a3722dec8bfee6a8c308d9574f0d6b3eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 11:07:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e23314a29a56c1ccc3eeb82eb39c666f26d92fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 11:08:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e16dc5d696e229bd531da67de93ebccd9bc30e2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 11:08:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"697b70c13d49098edc308c67d60cf308a044910e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:32:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94d37731d47d46aaa5e5603ffe3f5143 : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe666357ae56420dbbd914170560f219 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16043c98ab774302b51780b2afb95cae : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d70e25bf6648b189e2128a06b635d7 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2319f204b3f34479a33fbef2bda3cdb9 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7687fc33014f0aa7a6b2167cebc024 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4377388735a82afb90a39e405fd4970cbb23e8a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:33:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d1684404ff4ef42f9e76017cf28884db3674d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:55:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44efff9f6cfc40628a91dd9db9c0228f : SUCCESS in 1h 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd15562e60e497585a69ae53d59188a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfaefbbad05c4465a73df26fcf5bdf4c : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfddf5727b7046808c7b5c0e3d76ebd2 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c71ed1480e426992165187dd679815 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"5d0b76564b91c5478af51f899f78f81a2f34d872","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:55:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6049b087f7deaa833a6a8291315f59d70cf5b0bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 14:01:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c6d5441de3451dab4d82dea260d512 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"80bd6a5442a2503241dfc51dd533678accc02e52","revisions":{"3d6a47527b3bc5f859ef4546ec30c7b8226262c7":{"kind":"REWORK","_number":1,"created":"2021-03-23 11:53:43.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/40/782440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/40/782440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/40/782440/1"}}},"commit":{"parents":[{"commit":"406007e43eb48f959b25699423e9826b23ebe760","subject":"Uncap PrettyTable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/406007e43eb48f959b25699423e9826b23ebe760"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:00:14.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:00:26.000000000","tz":60},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: I402e708832395d95b04a5b956a19d5cc8c3f1c1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3d6a47527b3bc5f859ef4546ec30c7b8226262c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3d6a47527b3bc5f859ef4546ec30c7b8226262c7"}]},"branch":"refs/heads/master"},"9743e25d3477aa7ebab873f3cc9155f672d562cc":{"kind":"REWORK","_number":2,"created":"2021-04-07 14:39:10.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/40/782440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/40/782440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/40/782440/2"}}},"commit":{"parents":[{"commit":"406007e43eb48f959b25699423e9826b23ebe760","subject":"Uncap PrettyTable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/406007e43eb48f959b25699423e9826b23ebe760"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:00:14.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-07 14:46:20.000000000","tz":120},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: I402e708832395d95b04a5b956a19d5cc8c3f1c1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/9743e25d3477aa7ebab873f3cc9155f672d562cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/9743e25d3477aa7ebab873f3cc9155f672d562cc"}]},"branch":"refs/heads/master"},"80bd6a5442a2503241dfc51dd533678accc02e52":{"kind":"REWORK","_number":3,"created":"2021-04-16 11:07:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/782440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/40/782440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/40/782440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/40/782440/3"}}},"commit":{"parents":[{"commit":"406007e43eb48f959b25699423e9826b23ebe760","subject":"Uncap PrettyTable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/406007e43eb48f959b25699423e9826b23ebe760"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:00:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-16 11:07:10.000000000","tz":60},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version\neach 2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements [2] will stay aligned instead of relying on different\nversions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: I402e708832395d95b04a5b956a19d5cc8c3f1c1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/80bd6a5442a2503241dfc51dd533678accc02e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/80bd6a5442a2503241dfc51dd533678accc02e52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
