)]}'
{"id":"openstack%2Fautomaton~855722","triplet_id":"openstack%2Fautomaton~master~I868dcc578b25eb599acab5901a5dfc706071f5d4","project":"openstack/automaton","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:26:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I868dcc578b25eb599acab5901a5dfc706071f5d4","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:25:58.000000000","updated":"2022-09-07 09:26:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c75490d0caf753adf452fa22c5a818f4802deabc","_number":855722,"virtual_id_number":855722,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-03 22:12:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 22:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"546ca246a8246730d286cf94c331703f758abb00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3109ce71a8882e9b6de396be4994808bde11c4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 22:12:44.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ab8f9dc33464c9b82fe4c0c529b1d14\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8ed803c13164ff8b2fdbd1bb52e786d : SUCCESS in 1h 45m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2d2fd3e9074d64bbcd1281c0bb0d1a : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2277e78e2f0d43f98a25b653bb99bcaa : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e27945e208f4d0d8cafdceffb37a401 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936b40507cbc47aba7b5be868fb2cf13 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4cc16723864dbbaa81260d3da74c2d : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c75490d0caf753adf452fa22c5a818f4802deabc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:26:33.000000000","message":"Abandoned\n\nGate check passed (pep8 job failure is related to the health check patch), no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c26005261f82785877a9f330ed2a916c99efdde","revisions":{"0c26005261f82785877a9f330ed2a916c99efdde":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:25:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/22/855722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/22/855722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/855722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/855722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/855722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/22/855722/1"}}},"commit":{"parents":[{"commit":"0ea747e6bf3ab116d46882394346f094482cfd75","subject":"Fix formatting of release list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/0ea747e6bf3ab116d46882394346f094482cfd75"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:25:44.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:25:56.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I868dcc578b25eb599acab5901a5dfc706071f5d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/0c26005261f82785877a9f330ed2a916c99efdde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/0c26005261f82785877a9f330ed2a916c99efdde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
