)]}'
{"id":"openstack%2Fautomaton~873976","triplet_id":"openstack%2Fautomaton~master~I182b2e8f94795313a367fc0cec3344349524d25e","project":"openstack/automaton","branch":"master","topic":"release-health-check-cwi-antelope","hashtags":[],"change_id":"I182b2e8f94795313a367fc0cec3344349524d25e","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-02-15 19:26:28.000000000","updated":"2023-02-16 14:31:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e6d74233ef5dee6bb2ad8aa2abb2656bf6f4f31","_number":873976,"virtual_id_number":873976,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-15 20:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 20:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"593e84349555eebc3541da66c5dedad726eb7df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-15 19:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d263f50c54562da0f257dce17b31a1df317caeef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9116f56519e40c394a7a28719f466a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2ccb153410c4ad99244ea28ed5fc816 : SUCCESS in 1h 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb206c3356440a7bd54c8d92ca5f7da : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/848d65fff8b744b8ac9b3034751dd8e5 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa9e4127df4a46d28c21827cbb8fd6af : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9b27d4f850479bab994e6eb6896fd6 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2e6d74233ef5dee6bb2ad8aa2abb2656bf6f4f31","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-16 14:31:58.000000000","message":"Abandoned\n\nGate looks healthy, patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"171be813175456bfeabcda598109aba149c79b9e","revisions":{"171be813175456bfeabcda598109aba149c79b9e":{"kind":"REWORK","_number":1,"created":"2023-02-15 19:26:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/873976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/76/873976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/76/873976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/76/873976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/76/873976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/76/873976/1"}}},"commit":{"parents":[{"commit":"a4f76312250ae6855aabd4c726f9278ca59d5d33","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a4f76312250ae6855aabd4c726f9278ca59d5d33"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-15 19:26:19.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-15 19:26:26.000000000","tz":60},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I182b2e8f94795313a367fc0cec3344349524d25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/171be813175456bfeabcda598109aba149c79b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/171be813175456bfeabcda598109aba149c79b9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
