)]}'
{"id":"openstack%2Fautomaton~972296","triplet_id":"openstack%2Fautomaton~master~Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007","project":"openstack/automaton","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-09 13:29:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007","subject":"Add typing","status":"MERGED","created":"2026-01-06 14:31:07.000000000","updated":"2026-01-09 16:46:47.000000000","submitted":"2026-01-09 16:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972296-typing","meta_rev_id":"fd1c62000602c4d4923696093c4783d5dfa44025","_number":972296,"virtual_id_number":972296,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-09 16:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-09 13:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-09 13:29:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-06 16:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa904e58d848f99d4488ea23e279439800198858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-06 14:31:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ae63cbc61e3d679d61a13bcc62d090ac7224343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 16:32:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/621eea062a30487788b80369bc37853e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f47b14487049451891fa1265ad844fc7 : SUCCESS in 1m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb4d197a6ca44e1badfc1b259a0c5526 : FAILURE in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf3815a6acf64cac829f5f53dcc99f57 : SUCCESS in 1h 51m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a98e541d8d047be9bf5620a3233c25d : FAILURE in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90687e346f52477cbf5c0f7dd2f15cce : FAILURE in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acfacda61f2e4cfebf9dc51dbf3aba90 : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9554e5b8764df6ae4a843bb777be4f : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3d968ca2c754fc3abdf08036837b0c5 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e6b290fcdaddec3b1acd8f597dfcbf8fdf4930ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-07 10:15:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa4a91646a0def9169d188b08db8699ca6247beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-07 10:16:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d627619f10c767dc5c36805556da2d698fede0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:52:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79b6c62d1b6a48df962476cb61f1ace2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70a45ce07bf3430ea0983bf33983f5e8 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4089c060359540ae83bdf7b87d9256b1 : FAILURE in 2m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c1ec311cbc0436398e9883e95c5a1e2 : SUCCESS in 1h 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6844f8c6dd6b43cf8c94ca74d75d5ef1 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3371ca3f8c64cf58402d7fcc5fe3e1c : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4878b6bd687547f8bbc3657821048cee : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0051a55cb5404d81b9715126959c8d : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a7f1ba6bed43a1b55e5e75a1a8680d : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":3},{"id":"639725e62fb061de7f1de4a89eae10cefecaab84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-09 13:28:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"22c259327867e2815d2beeae1a3f34ab746bc178","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-09 13:29:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"068e3114070a3080930607bc893197a2e1b7e906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:38:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446047fa10954250b92caf5586ff8c50\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/89464ae8494245a69f4dc6d9d8995bbf : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f546dfc469b4a7fb9ae561d88edc9a2 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25a9bd326014dc18db4c64c60a87806 : SUCCESS in 1h 58m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8c56e799254ceeb1c80879fd8bf97b : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b670ba0302964dc99f6a024fbf838512 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ba1b39be8c144dbb6149ef14f83cae0 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337b176bb23641e18def674d651d0c50 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bf541a5e9054a07aa27a5575838eff3 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":4},{"id":"40efb795233b59d6a7b2260142e577eb0dd8d8c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:39:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d0be9027e4f7c990f8664c80eef612e2201cd2ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:45:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dced63b022a4b8980fa3e4e74bde2d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0442ea4c52a74dfc8c23dc93b508cdae : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a25be28d16a4263a019901ea5775d5f : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16db430db286475987b416c31042849d : SUCCESS in 1h 05m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef7fea64b3e4016b7968fdb55d86144 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09dbf8f0496d43dc8998fcffbdc53cef : SUCCESS in 1m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dc64bab4fab4d5cacffb6eb4c22883e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70cf25b2ced4e03bcb1a8f25cdf3dc3 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba7edacf315c4813aa4c6fbbf0bd194b : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"1d05aa20066b5e6ea4d71086f40107632bd187db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:45:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fd1c62000602c4d4923696093c4783d5dfa44025","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:46:47.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c10fa8a84444975b3417b2385c0f3ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f96e0c05b04d19b4c484ffc6f52072 : FAILURE in 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa65ab18f4b4ddb831540d7600af1f0 : FAILURE in 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b73bb3b67c83086909787b6262054a1688d313ec","revisions":{"ae334640b4554b9617bd27b72f93f101ec4bee29":{"kind":"REWORK","_number":1,"created":"2026-01-06 14:31:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/972296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/96/972296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/96/972296/1"}}},"commit":{"parents":[{"commit":"d4417df0b928f88806431d5b6cec453c271b72b4","subject":"trivial: Remove use of OrderedDict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d4417df0b928f88806431d5b6cec453c271b72b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-22 20:01:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-06 14:30:20.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/ae334640b4554b9617bd27b72f93f101ec4bee29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/ae334640b4554b9617bd27b72f93f101ec4bee29"}]},"branch":"refs/heads/master"},"56a24692ae98f76ebc627cd449ae13aea66c0dab":{"kind":"REWORK","_number":2,"created":"2026-01-07 10:15:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/972296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/96/972296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/96/972296/2"}}},"commit":{"parents":[{"commit":"d4417df0b928f88806431d5b6cec453c271b72b4","subject":"trivial: Remove use of OrderedDict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d4417df0b928f88806431d5b6cec453c271b72b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-22 20:01:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 10:14:08.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/56a24692ae98f76ebc627cd449ae13aea66c0dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/56a24692ae98f76ebc627cd449ae13aea66c0dab"}]},"branch":"refs/heads/master"},"6d01594254e48eb2f5a77489b68e506011539cf8":{"kind":"REWORK","_number":3,"created":"2026-01-07 10:16:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/972296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/96/972296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/96/972296/3"}}},"commit":{"parents":[{"commit":"d4417df0b928f88806431d5b6cec453c271b72b4","subject":"trivial: Remove use of OrderedDict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d4417df0b928f88806431d5b6cec453c271b72b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-22 20:01:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 10:16:35.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/6d01594254e48eb2f5a77489b68e506011539cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/6d01594254e48eb2f5a77489b68e506011539cf8"}]},"branch":"refs/heads/master"},"b73bb3b67c83086909787b6262054a1688d313ec":{"kind":"REWORK","_number":4,"created":"2026-01-09 13:28:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/972296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/96/972296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/96/972296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/96/972296/4"}}},"commit":{"parents":[{"commit":"d4417df0b928f88806431d5b6cec453c271b72b4","subject":"trivial: Remove use of OrderedDict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d4417df0b928f88806431d5b6cec453c271b72b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-22 20:01:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-09 13:28:53.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib34fab9de71cdadb0e4a8a211f86b9e6dc362007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/b73bb3b67c83086909787b6262054a1688d313ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/b73bb3b67c83086909787b6262054a1688d313ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
