)]}'
{"id":"openstack%2Fautomaton~979309","triplet_id":"openstack%2Fautomaton~master~I5a9fcdf9e65533a3bf9b2641be663ab51d19c5c6","project":"openstack/automaton","branch":"master","topic":"linter-sync","hashtags":[],"change_id":"I5a9fcdf9e65533a3bf9b2641be663ab51d19c5c6","subject":"ruff: Configure hacking as external linter","status":"MERGED","created":"2026-03-06 16:39:18.000000000","updated":"2026-03-07 13:37:47.000000000","submitted":"2026-03-07 13:36:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979309-linter-sync","meta_rev_id":"6eb5362df5af33def8be6d0fc1efa8bedae53488","_number":979309,"virtual_id_number":979309,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-07 13:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-07 11:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-07 11:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-06 18:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c6f1e78e6de332615ac0cd7c525fd99e99f427d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-06 16:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c769866674290db3b72e13172f9b9e60ea18af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:27:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30911cbdb19d4061b6f671c0c277702f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/264737f33ef641a7a3ae5ee2a55594e6 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10144228502c455f829132cb150d2668 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0670b6026c2b40199bba04e6e13ea3ee : SUCCESS in 1h 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a4bd5a0e7a46918f3b0da80701a45d : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e86eea2867044a6cab8528a99f1062fd : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0009c9b0e68847e999e48ea5a2a22711 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d83fef7713e41f2b3bd95853809119a : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714cd77b3e1742e8b79f11f0ac9075ee : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c19e949454120fd07516da8d25521b8870c57b43","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-07 11:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4e2db8486844d8da02a4027d9c9073374a171b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 11:51:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"689f51e4d977a0bf97f5780ec45f1d5043b4c28e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 13:36:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/178b69c71e78436bbc2d4d0e3b698b10\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9505619ca08b47ababe1932b73ae5e8a : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d535cae79ad64cd3856f2fa82fae175b : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08ece2e1f268493d933742c5f2947748 : SUCCESS in 1h 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abeea1fdb46411586651ce46c3f2e14 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83529c478d924f2fad6ab6201d32ec8f : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f92723f8c63446a7a2df383648a80704 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828837fbf35c43af961df56979b9f96e : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f96344f2524895d625691af853bb781140c4ec62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 13:36:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6eb5362df5af33def8be6d0fc1efa8bedae53488","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 13:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae977126de649c3a09583637dd84ab3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9ea611cccf42ee8f2b30a93cb860e5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb4d977a47d641b216bb8edde12a941ce53f0c87","revisions":{"fb4d977a47d641b216bb8edde12a941ce53f0c87":{"kind":"REWORK","_number":1,"created":"2026-03-06 16:39:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/979309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/09/979309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/09/979309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/09/979309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/09/979309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/09/979309/1"}}},"commit":{"parents":[{"commit":"7bb45f7827ea57ebd49093e07b9e0db6225df601","subject":"typing: Provide hint for reaction type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/7bb45f7827ea57ebd49093e07b9e0db6225df601"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-06 16:39:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-06 16:39:08.000000000","tz":0},"subject":"ruff: Configure hacking as external linter","message":"ruff: Configure hacking as external linter\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I5a9fcdf9e65533a3bf9b2641be663ab51d19c5c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/fb4d977a47d641b216bb8edde12a941ce53f0c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/fb4d977a47d641b216bb8edde12a941ce53f0c87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
