)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~444575","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ib70cea3887d8adbd67298138eb18f109605c9275","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"contraints","hashtags":[],"change_id":"Ib70cea3887d8adbd67298138eb18f109605c9275","subject":"Add Constraints support Adding constraints support to libraries is slightly more complex than services as the libraries themselves are listed in upper-constraints.txt which leads to errors that you can\u0027t install a specific version and a constrained version.","status":"ABANDONED","created":"2017-03-11 11:10:34.000000000","updated":"2021-02-16 13:25:08.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"752774f82cda9a86916f6e8d7939b251457fb198","_number":444575,"virtual_id_number":444575,"owner":{"_account_id":24924,"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","username":"Kiran_totad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},{"value":1,"date":"2017-03-15 04:19:49.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2017-03-15 14:35:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"value":-1,"date":"2021-02-11 11:32:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":-1,"date":"2017-03-17 09:12:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"},{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-11 11:11:18.000000000","updated_by":{"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},"reviewer":{"_account_id":20066,"name":"Soumiyajit","email":"soumiyajit@nectechnologies.in","username":"Soumiyajit"},"state":"REVIEWER"},{"updated":"2017-03-15 04:19:49.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-03-15 14:35:31.000000000","updated_by":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"reviewer":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"state":"REVIEWER"},{"updated":"2017-03-17 09:12:22.000000000","updated_by":{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"},"reviewer":{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"},"state":"REVIEWER"},{"updated":"2021-02-11 11:32:30.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"5687d1a07939fca3ce6294be2094e046a103d64c","author":{"_account_id":24924,"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","username":"Kiran_totad"},"date":"2017-03-11 11:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1d280881de3e1cb4daab8e53acc1beb0b5df4b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-11 11:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-barbican-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/75/444575/1/check/gate-barbican-tempest-plugin-pep8-ubuntu-xenial/9b55e8d/ : SUCCESS in 3m 43s\n- gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial-nv http://logs.openstack.org/75/444575/1/check/gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial-nv/01ad7c0/ : SUCCESS in 38m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f81bf7eafeed5e34cd1776fa3c69fb5348a00714","author":{"_account_id":16034,"name":"soumiyajit","email":"somya.rock@gmail.com","username":"soumiyajit"},"date":"2017-03-13 17:59:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"337adde60f8c8e68b837409a561f2e6761442afc","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-03-15 00:45:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThe \"+\" characters from a diff have made it into tools/tox_install.sh. Please fix.","accounts_in_message":[],"_revision_number":1},{"id":"9be6e533aee97cd1254cdde1afffc777493622b8","author":{"_account_id":24924,"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","username":"Kiran_totad"},"date":"2017-03-15 03:58:57.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"a67b273e09cd297e89855c4b07cac61750c0cdf0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-15 04:19:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-barbican-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/75/444575/2/check/gate-barbican-tempest-plugin-pep8-ubuntu-xenial/a3f810f/ : SUCCESS in 3m 20s\n- gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial-nv http://logs.openstack.org/75/444575/2/check/gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial-nv/54c3f63/ : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"501389ac83f5991ad1c3a87feac10a5519f252fd","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-03-15 14:35:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI\u0027m unsure what you\u0027re proposing to do with this patch. Doesn\u0027t the use of the `-c` flag in the `pip install` line in tox.ini satisfy what tox_install.sh is attempting to do? Also, it doesn\u0027t look like tox_install.sh is ever being accessed.","accounts_in_message":[],"_revision_number":2},{"id":"b9a45d8f90a73d4edfba2c0d61875ca5c730f353","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-03-16 01:20:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"828ad75b67eae53c4dd17e7cbd01815536873c94","author":{"_account_id":20863,"name":"Sharat Sharma","email":"sharat.sharma@in.ibm.com","username":"sharat.sharma"},"date":"2017-03-17 09:12:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13d4c48dd3f7a9b09cbdc407cdab566e639b2aee","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-11 11:32:30.000000000","message":"Patch Set 2: Code-Review-1\n\nI guess it\u0027s really time to abandon this change","accounts_in_message":[],"_revision_number":2},{"id":"752774f82cda9a86916f6e8d7939b251457fb198","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-02-16 13:25:08.000000000","message":"Abandoned\n\nAbandoned per Luigi\u0027s comments.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06bc219bca05e5d6a952021461d2fb87f9f6f056","revisions":{"a92392bbde9f8aea9be192dbbb3786478c3d6a6b":{"kind":"REWORK","_number":1,"created":"2017-03-11 11:10:34.000000000","uploader":{"_account_id":24924,"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","username":"Kiran_totad"},"ref":"refs/changes/75/444575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/75/444575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/1"}}},"commit":{"parents":[{"commit":"5af45595bb113cd39a40ae647dae252befe0aadf","subject":"Add image signing scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5af45595bb113cd39a40ae647dae252befe0aadf"}]}],"author":{"name":"Kiran_totad","email":"kiran.totad@nectechnologies.in","date":"2017-03-11 10:14:13.000000000","tz":330},"committer":{"name":"Kiran_totad","email":"kiran.totad@nectechnologies.in","date":"2017-03-11 10:15:57.000000000","tz":330},"subject":"Add Constraints support Adding constraints support to libraries is slightly more complex than services as the libraries themselves are listed in upper-constraints.txt which leads to errors that you can\u0027t install a specific version and a constrained version.","message":"Add Constraints support\nAdding constraints support to libraries is slightly more\ncomplex than services as the libraries themselves are listed\nin upper-constraints.txt which leads to errors that you can\u0027t\ninstall a specific version and a constrained version.\n\nThis change adds constraints support by also adding a helper\nscript to edit the constraints to remove barbican-tempest-plugin.\n\nChange-Id: Ib70cea3887d8adbd67298138eb18f109605c9275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a92392bbde9f8aea9be192dbbb3786478c3d6a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a92392bbde9f8aea9be192dbbb3786478c3d6a6b"}]},"branch":"refs/heads/master"},"06bc219bca05e5d6a952021461d2fb87f9f6f056":{"kind":"REWORK","_number":2,"created":"2017-03-15 03:58:57.000000000","uploader":{"_account_id":24924,"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","username":"Kiran_totad"},"ref":"refs/changes/75/444575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/75/444575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/75/444575/2"}}},"commit":{"parents":[{"commit":"5af45595bb113cd39a40ae647dae252befe0aadf","subject":"Add image signing scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5af45595bb113cd39a40ae647dae252befe0aadf"}]}],"author":{"name":"Kiran_totad","email":"kiran.totad@nectechnologies.in","date":"2017-03-11 10:14:13.000000000","tz":330},"committer":{"name":"Kiran Totad","email":"kiran.totad@nectechnologies.in","date":"2017-03-15 03:58:45.000000000","tz":0},"subject":"Add Constraints support Adding constraints support to libraries is slightly more complex than services as the libraries themselves are listed in upper-constraints.txt which leads to errors that you can\u0027t install a specific version and a constrained version.","message":"Add Constraints support\nAdding constraints support to libraries is slightly more\ncomplex than services as the libraries themselves are listed\nin upper-constraints.txt which leads to errors that you can\u0027t\ninstall a specific version and a constrained version.\n\nThis change adds constraints support by also adding a helper\nscript to edit the constraints to remove barbican-tempest-plugin.\n\nChange-Id: Ib70cea3887d8adbd67298138eb18f109605c9275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/06bc219bca05e5d6a952021461d2fb87f9f6f056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/06bc219bca05e5d6a952021461d2fb87f9f6f056"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
