)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~481440","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ia690ea0a284f1e1cfe7e223613acdb30e40a9980","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"tempest","hashtags":[],"change_id":"Ia690ea0a284f1e1cfe7e223613acdb30e40a9980","subject":"Remove old pep8 ignores that are no longer necessary","status":"MERGED","created":"2017-07-07 05:54:36.000000000","updated":"2021-05-25 14:02:31.000000000","submitted":"2021-05-25 14:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"481440-tempest","meta_rev_id":"c3983224f65a0d665bc401c55baf3395e70592b5","_number":481440,"virtual_id_number":481440,"owner":{"_account_id":26271,"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","username":"arundhati.surpur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 14:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-05-25 13:30:15.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},{"value":2,"date":"2021-05-18 13:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-05-25 13:30:15.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-07 06:27:13.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-25 08:59:05.000000000","updated_by":{"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},"reviewer":{"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},"state":"REVIEWER"},{"updated":"2021-02-11 11:35:52.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-03-02 13:39:13.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-03-02 14:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 13:13:59.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-05-25 13:30:15.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"09d95103fe174f05dacacaedcf3c88379b905311","author":{"_account_id":26271,"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","username":"arundhati.surpur"},"date":"2017-07-07 05:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9ffa6f47aa60f673df437cdf1431bd52006eb75","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 06:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-barbican-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/40/481440/1/check/gate-barbican-tempest-plugin-pep8-ubuntu-xenial/99db1f2/ : SUCCESS in 3m 22s\n- gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial http://logs.openstack.org/40/481440/1/check/gate-barbican-simple-crypto-dsvm-tempest-ubuntu-xenial/d7067d2/ : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8a815784ca6289107dc89c2d2a8c0179ec471206","author":{"_account_id":21797,"name":"Jeremy Liu","display_name":"Jeremy Liu","email":"liujiong.cc@gmail.com","username":"liujiong"},"date":"2017-07-25 08:59:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nIt would be better if you add a brief description in commit message about what E123 E125 do","accounts_in_message":[],"_revision_number":1},{"id":"f62fbf2090d39d6871fcb9bbd9b9997e9b22c0b6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-11 11:35:52.000000000","message":"Patch Set 1:\n\nThis needs a rebase, but it seems still relevant","accounts_in_message":[],"_revision_number":1},{"id":"fe3753fe8f23dfbe9b5d4e705d5aa69e4de078f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-02 13:39:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53ee484012e798d0f8c372761035bbb57581e923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:21:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22968fca45474b67baf600a1120a236c : SUCCESS in 3m 41s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/114b795b3c6e42d58c0c97934898e148 : SUCCESS in 40m 41s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/98b7259c0f8242678f106fcbad7d7935 : SUCCESS in 36m 08s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/8c7a74958b60440d807210d5183c485c : SUCCESS in 36m 14s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/985fec12ce6f463c869060fb29b495d7 : SUCCESS in 36m 24s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a72a1002267b4a6ea3af61c4e2c87c70 : SUCCESS in 30m 54s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/5c59f4b46d0a4b619a9e94aaea863e2e : SUCCESS in 40m 19s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/20f371ced525467e9c9a807642f5e306 : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a2e284644d06e5a8d3a9984594ade0cbe179deae","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-05-18 13:13:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2a0accb0f7def5237d3f4df136723f403d76df6","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-05-25 13:30:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d03bef5e3490cda1981f5a0a8218aef30119c5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:30:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9dd6940512e47a6125b61260961a611e3f5ebf0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:02:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f719193198448bf9597cd77628faf66 : SUCCESS in 3m 35s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3557de38689341c790a76709ce7fb997 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c3983224f65a0d665bc401c55baf3395e70592b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:02:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42eec762be9b7ca608487c207bd80f78ea1b535c","revisions":{"92cf8fbc74b990a6eb2745796f3b942d639de843":{"kind":"REWORK","_number":1,"created":"2017-07-07 05:54:36.000000000","uploader":{"_account_id":26271,"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","username":"arundhati.surpur"},"ref":"refs/changes/40/481440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/40/481440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/1"}}},"commit":{"parents":[{"commit":"f2055253e95f240733ea12de1d93b6f8d3208765","subject":"Merge \"Fix tox command for running tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2055253e95f240733ea12de1d93b6f8d3208765"}]}],"author":{"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","date":"2017-07-07 05:54:34.000000000","tz":330},"committer":{"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","date":"2017-07-07 05:56:05.000000000","tz":330},"subject":"Remove old pep8 ignores that are no longer necessary","message":"Remove old pep8 ignores that are no longer necessary\n\nChange-Id: Ia690ea0a284f1e1cfe7e223613acdb30e40a9980\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/92cf8fbc74b990a6eb2745796f3b942d639de843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/92cf8fbc74b990a6eb2745796f3b942d639de843"}]},"branch":"refs/heads/master"},"42eec762be9b7ca608487c207bd80f78ea1b535c":{"kind":"REWORK","_number":2,"created":"2021-03-02 13:39:13.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/40/481440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/40/481440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/481440/2"}}},"commit":{"parents":[{"commit":"ad7f742ea9edd5508123b321d46ec2f4e1af8271","subject":"Merge \"Fix test_signed_image_upload_boot_failure skip method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ad7f742ea9edd5508123b321d46ec2f4e1af8271"}]}],"author":{"name":"Arundhati Surpur","email":"arundhati@nectechnologies.in","date":"2017-07-07 05:54:34.000000000","tz":330},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-03-02 13:38:08.000000000","tz":60},"subject":"Remove old pep8 ignores that are no longer necessary","message":"Remove old pep8 ignores that are no longer necessary\n\nChange-Id: Ia690ea0a284f1e1cfe7e223613acdb30e40a9980\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/42eec762be9b7ca608487c207bd80f78ea1b535c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/42eec762be9b7ca608487c207bd80f78ea1b535c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
