)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~515210","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ica1d32ca66d6f89692e9cc90e415bf569332f58d","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"bp/nova-validate-certificates","hashtags":[],"change_id":"Ica1d32ca66d6f89692e9cc90e415bf569332f58d","subject":"Add certificate validation scenario tests","status":"MERGED","created":"2017-10-25 21:10:36.000000000","updated":"2020-08-25 11:01:01.000000000","submitted":"2020-08-25 11:01:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"515210-1598353261587-187197d0","meta_rev_id":"1867292535b4cd138a35754f67f55dab5c93e5e2","_number":515210,"virtual_id_number":515210,"owner":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 11:01:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"recommended":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2020-08-25 10:09:58.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2020-03-13 21:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2020-08-25 10:09:58.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},{"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-25 21:22:39.000000000","updated_by":{"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},"reviewer":{"_account_id":7764,"name":"Peter Hamilton","email":"peter.hamilton@jhuapl.edu","username":"Peter"},"state":"REVIEWER"},{"updated":"2017-10-25 21:22:48.000000000","updated_by":{"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},"reviewer":{"_account_id":8623,"name":"Kaitlin Farr","email":"kaitlin.farr@jhuapl.edu","username":"kaitlin.farr"},"state":"REVIEWER"},{"updated":"2017-12-16 01:52:53.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2018-01-08 13:57:59.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-04-13 16:10:39.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2018-04-16 21:35:58.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-06-04 14:23:58.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-06-22 16:20:25.000000000","updated_by":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"reviewer":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"state":"REVIEWER"},{"updated":"2018-06-29 12:54:11.000000000","updated_by":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2018-06-29 12:54:48.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-08-06 16:33:48.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2019-08-14 15:47:33.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-09-26 06:39:58.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-10-12 08:13:38.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-13 14:39:28.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-13 15:00:03.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-10-31 05:41:52.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2020-08-25 10:09:58.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-08-25 11:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52109df73180c39b4758848d4871b3948d658494","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-10-25 21:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73a954374de97d945f4ed7c6ff700a68e0cc93e8","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-10-25 21:10:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4ac32cc67455d0596b88c70f98b9598583022bbb","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-10-25 21:21:43.000000000","message":"Patch Set 1:\n\nStill WIP because we\u0027re currently running into an issue with passing the trusted_image_certificates in the create_test_server call:\n\nDEBUG nova.api.openstack.wsgi [None req-13f7aa31-013d-41d9-95bf-03634a302ef2 tempest-ImageSigningTest-1588901414 tempest-ImageSigningTest-1588901414] Returning 400 to user: Invalid input for field/attribute server. Value: {u\u0027trusted_image_certificates\u0027: [u\u00279a40caaf-54af-4b97-86fe-b4080eeb6061\u0027], u\u0027flavorRef\u0027: u\u00271\u0027, u\u0027name\u0027: u\u0027tempest-ImageSigningTest-server-634016441\u0027, u\u0027imageRef\u0027: u\u00272cc8c5a3-4715-49eb-bf63-81f5ffb4a36e\u0027}. Additional properties are not allowed (u\u0027trusted_image_certificates\u0027 was unexpected) {{(pid\u003d13157) __call__ /opt/stack/nova/nova/api/openstack/wsgi.py:1029}}","accounts_in_message":[],"_revision_number":1},{"id":"4ea656aafcdac47685382c03b2bcb5248e9c0be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 21:53:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/1/check/openstack-tox-pep8/ca5703f/ : SUCCESS in 5m 07s\n- legacy-barbican-simple-crypto-dsvm-tempest http://logs.openstack.org/10/515210/1/check/legacy-barbican-simple-crypto-dsvm-tempest/2ea93fb/ : FAILURE in 38m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f355d4d85967b5c6c5d988a75a8762599a1d5a18","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-10-25 22:06:09.000000000","message":"Patch Set 1:\n\ntrusted_image_certificates parameter needs to be specified in all calls to create_test_server (i.e., test_volume_encryption and test_ephemeral_disk_encryption)","accounts_in_message":[],"_revision_number":1},{"id":"91c231c531545b38bda73c263d90acb58ea5c657","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-10-25 23:05:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cc876fc9461029541e9af92298591c76673b26a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 23:47:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/2/check/openstack-tox-pep8/9779af3/ : SUCCESS in 5m 26s\n- legacy-barbican-simple-crypto-dsvm-tempest http://logs.openstack.org/10/515210/2/check/legacy-barbican-simple-crypto-dsvm-tempest/98c7764/ : FAILURE in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b26bef65da16bf49b915fcdd556641ac7f5762a1","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-11-30 20:25:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"104df0e4cd4b77aebd9537075c3c404a9098a8d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 20:25:48.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6e91c68f884216fe902b526c0ae851bb94b36cbd","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-11-30 20:26:27.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3df00be4fb06b79482fe29067453b30195e19c00","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-01 20:49:32.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a74efb997aae2004566af55b98ed04b11dc7794b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-01 20:49:57.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"176afcba836cbba9bd2f7b67be7cb396b927b473","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-01 20:50:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"741344059766369e83d064d25afc08500061d130","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-05 20:52:03.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d35d09140137581f87d6e5c1f0481f0b7eecc64b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 20:52:22.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d5feb1f50ab64d94970bb59e430d6c1f81611f7e","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-05 20:54:44.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"fdfe87edfd371e0e346a24ea83b58ea6857d6fc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 21:27:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/6/check/openstack-tox-pep8/2296552/ : SUCCESS in 3m 45s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/6/check/barbican-simple-crypto-devstack-tempest/09fc641/ : FAILURE in 28m 57s","accounts_in_message":[],"_revision_number":6},{"id":"3ef4c525637ff296f10b0dac1c3dec1c0b5ba6cf","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-05 21:43:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c33069d30846d744be65e2e9a68f4ffaa1835236","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-05 21:56:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8c6fb4b46bd95e8204b349bab42d13fe2312b37d","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-05 22:24:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"07270c5a1e26384ad3a22eb851becda472c76bb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 22:56:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/7/check/openstack-tox-pep8/f167821/ : SUCCESS in 3m 21s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/7/check/barbican-simple-crypto-devstack-tempest/af0daff/ : FAILURE in 28m 06s","accounts_in_message":[],"_revision_number":7},{"id":"1402ed292270f30a6b362742178c57efa20aa295","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-06 18:46:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fa5123da0ca0a725ae00837ac9f6662a94bfc0e8","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-07 19:26:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a7fdf5cd1453fb84a98807edd5496f82cbffb13e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 20:03:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/8/check/openstack-tox-pep8/7745538/ : SUCCESS in 5m 58s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/8/check/barbican-simple-crypto-devstack-tempest/cc592f7/ : FAILURE in 33m 46s","accounts_in_message":[],"_revision_number":8},{"id":"f78287eca35505c992802ff429fa78937b1f4b56","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-07 22:33:17.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"93fb272b73f57b100b9fc370f66a39d18d61c9cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 23:09:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/9/check/openstack-tox-pep8/21a19dc/ : SUCCESS in 4m 38s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/9/check/barbican-simple-crypto-devstack-tempest/a888552/ : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":9},{"id":"359a9d4314c51b9ab873bdd1fd7d974af0a60dd3","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-08 18:02:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a62cf9a6856d81c51d6c07c6cc725b0b2efdf9ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 18:04:56.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"d011be90fdbb9a1cf6e7e1af23730e494468ce80","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-08 18:34:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"efb6325d2b069f4a90d566f1c6733857e2f3f7ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 19:28:32.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/11/check/openstack-tox-pep8/c4808ab/ : FAILURE in 4m 15s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/11/check/barbican-simple-crypto-devstack-tempest/619db62/ : FAILURE in 28m 56s","accounts_in_message":[],"_revision_number":11},{"id":"c37a82093b3fe852d7d85936a41dcfbd52bdd6b8","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-08 21:24:16.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nIt looks like the API version is now set to 2.57 successfully: http://logs.openstack.org/10/515210/11/check/barbican-simple-crypto-devstack-tempest/619db62/logs/screen-n-api.txt.gz#_Dec_08_19_26_06_679296","accounts_in_message":[],"_revision_number":11},{"id":"8d8446e3dd6ee16de1d118f1b5b26c4239e9bba2","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-08 22:30:55.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5813651f2065ebcd366f24d19e37b1ac540aa93f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 23:03:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/12/check/openstack-tox-pep8/a03e09c/ : SUCCESS in 4m 20s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/12/check/barbican-simple-crypto-devstack-tempest/2bca6c3/ : FAILURE in 28m 47s","accounts_in_message":[],"_revision_number":12},{"id":"de305ed579a434bb13691b49dde94280cd4c1578","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-08 23:51:59.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"7481f56dd9a02fb4771363ae5a86a14195cf7705","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 00:22:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/13/check/openstack-tox-pep8/5761ee6/ : SUCCESS in 3m 39s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/13/check/barbican-simple-crypto-devstack-tempest/b185381/ : FAILURE in 27m 03s","accounts_in_message":[],"_revision_number":13},{"id":"1b5d84c3a0a247ac3c04fc0ece9549ec8c0bcdc3","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-09 03:55:38.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"96eac853f2497786b3293f35b052f577d941c8fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 04:32:28.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/14/check/openstack-tox-pep8/1636412/ : SUCCESS in 5m 42s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/14/check/barbican-simple-crypto-devstack-tempest/3ba0ce9/ : FAILURE in 33m 58s","accounts_in_message":[],"_revision_number":14},{"id":"4a2bd5226bb2eae942f015708d911e707b18ba94","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-11 19:51:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"89081db75f20b2bef7dafda0330e907f88333e69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-11 20:39:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/15/check/openstack-tox-pep8/bfabf22/ : SUCCESS in 3m 24s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/15/check/barbican-simple-crypto-devstack-tempest/7d1973a/ : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":15},{"id":"86c86c86f4e3d9df16043feb865594b139a824ad","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-11 21:11:56.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nIt looks like there are two types of failures (see http://logs.openstack.org/10/515210/15/check/barbican-simple-crypto-devstack-tempest/7d1973a/logs/testr_results.html.gz )\n\n1. The resource could not be found -- this appears to be security_groups. If you remove the \"security_group \u003d self._create_security_group()\" lines and the \"security_groups\u003d[{\u0027name\u0027: security_group[\u0027name\u0027]}],\" lines, this error shouldn\u0027t happen.\n\n2. Could not locate the signing certificate for the base certificate in the set of trusted certificates.\nThis is because the certificate isn\u0027t configured properly (see a clearer error at: http://logs.openstack.org/10/515210/15/check/barbican-simple-crypto-devstack-tempest/7d1973a/logs/screen-n-cpu.txt.gz#_Dec_11_20_37_28_823271 )","accounts_in_message":[],"_revision_number":15},{"id":"02c41626f3b0f1230095abdf3bba7f0e041008ad","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-11 21:33:20.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nThanks for the help with parsing the logs, @Brianna. Is there no longer a need to create a test server with a test security group?","accounts_in_message":[],"_revision_number":15},{"id":"61474c5ba8a0c55932f45806bf3f434564d20626","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-11 21:36:44.000000000","message":"Patch Set 15:\n\n@Jackie: I don\u0027t think the security group is needed anymore, since it\u0027s been deprecated since microversion 2.36: https://github.com/openstack/nova/blob/master/api-ref/source/os-security-groups.inc","accounts_in_message":[],"_revision_number":15},{"id":"0e5ecab3c407ec7f2483dfab80f81bbc958f5137","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-11 21:55:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a4d002227b7d5e3193b25532c0467c4d28b16f5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-11 22:45:49.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/16/check/openstack-tox-pep8/0b0b52e/ : SUCCESS in 7m 10s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/16/check/barbican-simple-crypto-devstack-tempest/38c019e/ : FAILURE in 41m 54s","accounts_in_message":[],"_revision_number":16},{"id":"3ded3ed7b6e3faa355a8ec290290e453a01a8493","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-12 15:16:54.000000000","message":"Patch Set 16:\n\nIt looks like floating IPs are also deprecated [1][2]. Is there any reason barbican-tempest-plugin uses its own manager.py, rather than just importing from tempest.scenario?\n\n[1] http://logs.openstack.org/10/515210/16/check/barbican-simple-crypto-devstack-tempest/38c019e/job-output.txt.gz#_2017-12-11_22_40_28_218112\n[2] https://github.com/openstack/nova/blob/master/api-ref/source/os-floating-ips.inc","accounts_in_message":[],"_revision_number":16},{"id":"158cc32057994eb0cc02e306923aa986a5bb2a90","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-12 15:34:08.000000000","message":"Patch Set 16:\n\n@Jackie: All tempest plugins were instructed to copy manager.py locally while refactoring took place in tempest.  Now that the refactoring is complete, barbican-tempest-plugin should have its own manager.py deleted, and import directly from tempest instead.","accounts_in_message":[],"_revision_number":16},{"id":"41f575d3404d7e2ddeb862db041202d71ed3620a","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-12 16:28:04.000000000","message":"Patch Set 16:\n\n@Brianna: Can do. Is it acceptable to do that in this patch, or should we create a separate patch for that transition?\n\nIt doesn\u0027t look like the deprecation of the proxy APIs (security groups [1], floating IPs, and images [3]) is reflected in tempest\u0027s latest manager.py [4]. Removing the creation and use of security groups eliminates the \"Object not found\" error for security_groups, but the same error still occurs for floating IPs and compute_images_client [5] and is much more tedious to remediate.\n\n[1] https://github.com/openstack/nova/blob/master/api-ref/source/os-security-groups.inc\n[2] https://github.com/openstack/nova/blob/master/api-ref/source/os-floating-ips.inc\n[3] https://github.com/openstack/nova/blob/master/api-ref/source/images.inc\n[4] https://github.com/openstack/tempest/blob/master/tempest/scenario/manager.py\n[5] https://github.com/openstack/barbican-tempest-plugin/blob/master/barbican_tempest_plugin/tests/scenario/test_image_signing.py#L76","accounts_in_message":[],"_revision_number":16},{"id":"c327c585d197c2fc8bb12ef507c6f9d4bcd93c99","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-12 16:35:24.000000000","message":"Patch Set 16:\n\n@Jackie: I think a separate patch should be created to remove the local manager.py\n\nAs for the other deprecated options, would it be possible to use the default API (2.1) for all of the tests except for the specific certificate validation tests that require 2.57?","accounts_in_message":[],"_revision_number":16},{"id":"2412129cbf8d17173ec1b8c07529c715e16824c4","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-12 18:04:57.000000000","message":"Patch Set 16:\n\n@Brianna: Yes, it is possible to have all of the tests except the tests specific to certificate validation use API microversion 2.1. We would need to specify it at the top of each test class, since barbican_manager has been modified to retrieve the API microversion from each class (specified as `min_microversion`).\n\nHowever, I think using different microversions like this would be problematic: For this test, we need to enable certificate validation in the nova configurations. This means that any attempts to create a test server without specifying trusted image certificates would fail. At the same time, since API 2.1 does not support the trusted_image_certificates parameter, we are not able to specify the trusted image certificates...","accounts_in_message":[],"_revision_number":16},{"id":"f717485f0e6c835d822dd5c3f11c846c874073ef","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-12 18:43:00.000000000","message":"Patch Set 16:\n\n@Jackie: I think there\u0027s a way to get around the other tests failing.\n\nSee the nova spec at: https://specs.openstack.org/openstack/nova-specs/specs/queens/approved/nova-validate-certificates.html\n\n\"It is possible a silent fail scenario could occur if (1) nova is not configured to conduct certificate validation, and (2) the user provides trusted certificate IDs expecting certificate validation to occur. In this case, nova would not conduct certificate validation and would boot the instance, causing the user to believe certificate validation succeeded even though it never happened. To prevent this from happening, the boot workflow will be updated to conduct signature and certificate verification if trusted certificate IDs are associated with the instance data. This matches the user’s expectations and prevents a silent fail scenario from ever occurring. Note here that this override only occurs if the user specifies certificate IDs. The default list of certificate IDs is only used if the feature is enabled and therefore will never trigger the override.\"\n\nWe could set enable_certificate_validation to false, but the cert validation should still occur as long as we provide the certificate ids.  We can confirm that it happens by providing both a correctly signed certificate and an incorrectly-signed certificate, and show that the latter fails while the former succeeds.","accounts_in_message":[],"_revision_number":16},{"id":"591515bb191b73ad92c05fe43fa5a4df46da8f3f","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-12 19:06:06.000000000","message":"Patch Set 16:\n\n@Brianna: I don\u0027t think the current state of the certificate validation patches will allow that. We need to update patch 479949 to validate certificates when trusted_certs are supplied, rather than checking the enable_certificate_validation configuration.\n\n[1] https://review.openstack.org/#/c/479949/22/nova/image/glance.py","accounts_in_message":[],"_revision_number":16},{"id":"13ad65f1d66c9c3d51a9b9f0cbea8861a1ea7e76","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-12 19:32:45.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)\n\n@Jackie: Agreed, an update is needed to that patch.  It should perform certificate validation if the enable_certificate_validation is True and/or if trusted_certs is provided.\n\nI determined that the \"bad signing certificate\" is actually not invalid, and should be modified so that cursive won\u0027t identify it as a valid signing certificate.","accounts_in_message":[],"_revision_number":16},{"id":"9701c29d72ace4fe70756288556871a8ae3bca87","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-12 22:35:33.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a0991bf69862e07688cec82210a6b92baca613e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-12 22:36:42.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"e1848179da3693ac860f3a21f9270feabc0ebbf6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-12 22:44:37.000000000","message":"Patch Set 17:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"69175eddc3957853eecf5b8738059aff9fc90f60","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-13 02:04:16.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"800d741f767ff8f30a231e07902e8c391018c2e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 02:05:14.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"bfc05b0ce1cc6e2da76f07ed44eb5bf95732ee03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-13 15:18:07.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"2414dd7e94eae285248c07a38c9082b784007c23","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-13 15:55:44.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"e22bee186c50bab333d66f5e68a3b67604b6a349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 17:01:30.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/19/check/openstack-tox-pep8/cb3aa76/ : SUCCESS in 4m 35s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/19/check/barbican-simple-crypto-devstack-tempest/e326c0a/ : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":19},{"id":"2e646e64befeeeced7abdcd549a8e7e1720efc45","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2017-12-13 17:17:09.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"782fa7524faa028bc4b5397aa2a70ddd87686754","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-13 21:11:02.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"6a3b21a918bead8e37b059f901373f30bab68dfa","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-13 21:51:47.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"f8d14f42e34ce7f17a860f533dc280a812525ec0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 21:52:15.000000000","message":"Patch Set 20: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":20},{"id":"fe7a49a141f33d8bda2caef6ffcbc749cf7eb67f","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-14 16:38:38.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6624bd3999aac3bd7caa014eed2efdf19358f093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 16:39:06.000000000","message":"Patch Set 21: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"a96edef216f8d030f9dd2027b24b82b2044ad611","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-14 19:51:17.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"98f0498360099e77f62a2901be8ce3f3b3f31317","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-16 00:27:43.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"f3d7bdd1cdc0a0852ca08f32b1362ba56ac06f87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-16 00:59:02.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/22/check/openstack-tox-pep8/55600f8/ : SUCCESS in 4m 15s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/22/check/barbican-simple-crypto-devstack-tempest/9b35317/ : FAILURE in 28m 03s","accounts_in_message":[],"_revision_number":22},{"id":"fa210c8277c83777ffadc80d64446ca2dc73a6d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-16 01:52:53.000000000","message":"Patch Set 22:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":22},{"id":"c644265dc5c035588f0b946a89b5119200400934","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-16 02:40:10.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"f6a9056a9cf3fb185e29d4720cfa7868411d51e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-16 03:17:05.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/23/check/openstack-tox-pep8/75ea43c/ : SUCCESS in 6m 13s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/23/check/barbican-simple-crypto-devstack-tempest/9396143/ : FAILURE in 31m 06s","accounts_in_message":[],"_revision_number":23},{"id":"8f9bad3078f011247d5f72992bded4d8591c8a44","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2017-12-16 20:38:13.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"6f9336de7b2e192a765fce850dfb99db310a7947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-16 21:11:59.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/24/check/openstack-tox-pep8/cab2432/ : SUCCESS in 3m 22s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/24/check/barbican-simple-crypto-devstack-tempest/70e70a4/ : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":24},{"id":"a5a35de80f0627c23056ff30879999ec7a983316","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-01-04 20:46:02.000000000","message":"Patch Set 24:\n\nAs part of the nova-validate-certificates blueprint, this patch is tangled up with a bunch of other patches. In an attempt to lower the overhead needed to review these patches, we have created an etherpad with usage and testing instructions:\n\nhttps://etherpad.openstack.org/p/queens-nova-certificate-validation","accounts_in_message":[],"_revision_number":24},{"id":"aa922e2df4a125f7c60bae6399075717031aabe5","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-01-22 15:22:47.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"f305de476a93312afd38d256077e4c72cdc5ca61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 16:07:22.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/25/check/openstack-tox-pep8/2dffbcd/ : SUCCESS in 3m 29s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/25/check/barbican-simple-crypto-devstack-tempest/c355485/ : FAILURE in 32m 44s","accounts_in_message":[],"_revision_number":25},{"id":"6b38129a458392ebcd7f36e8360ac4b3df6909dc","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-01-22 16:11:44.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"6572a37f0697dc446f3a2ab00e4a8fc658ecb27f","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-01-22 16:12:47.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"53a961b26c76ece368dfe9cc66d344a9f2a11ab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 17:18:39.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/26/check/openstack-tox-pep8/9cdbe0a/ : SUCCESS in 4m 54s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/26/check/barbican-simple-crypto-devstack-tempest/3159aef/ : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":26},{"id":"1ba85c9d9aa9043cb6f4ab17a955d87b1f84344e","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-01-23 15:12:34.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"6668ca35f313f2059356b5a2430d4bcaf6984365","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 16:52:16.000000000","message":"Patch Set 27: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":27},{"id":"2477e35c48f30315f82715ad6dd6d3514d57be45","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-01-23 16:59:00.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"50eae41071b8dc5f818835dc3876e7e5b8282214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 19:19:43.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/27/check/openstack-tox-pep8/8486175/ : SUCCESS in 4m 11s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/27/check/barbican-simple-crypto-devstack-tempest/4f5952c/ : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":27},{"id":"c6586a98f2a61cc5a1e420254037af46109801d4","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-01-25 13:33:49.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"3dcc389f5e6f4b2dbd09f2f460385933a84d3ac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-25 15:18:39.000000000","message":"Patch Set 27: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":27},{"id":"dd99a7aeb97a1f3f24467d260837f7e457a73ae4","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-01-25 22:24:53.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"1ce291656f29b4067fc1d702a3147ffadbf2d54c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-25 23:44:03.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/27/check/openstack-tox-pep8/5fdd495/ : SUCCESS in 4m 14s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/27/check/barbican-simple-crypto-devstack-tempest/08dcdef/ : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":27},{"id":"aa9d91d19d34aceb37ef917e4e96151b63731f98","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-03-13 19:48:29.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"aba1bdd9de3700db137a08fc05f20ecad8d8b459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-13 21:25:49.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/27/check/openstack-tox-pep8/7acd480/ : SUCCESS in 5m 00s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/27/check/barbican-simple-crypto-devstack-tempest/388f519/ : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":27},{"id":"80e6676b538b07e021b21af494230a5981b9fc7d","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-03-22 19:05:29.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"294a8f384ea54b4c689f548ea37569f3d7297671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 20:10:39.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/28/check/openstack-tox-pep8/38b0ec1/ : FAILURE in 4m 02s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/28/check/barbican-simple-crypto-devstack-tempest/acea0e4/ : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":28},{"id":"faf605a385b8efe78524ee6c5e85d0688cfcae8f","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-03-23 13:35:07.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8a005856a17142b81a6b44ab02032d84b91857a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 14:11:37.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/d4e3b06/ : SUCCESS in 3m 30s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/0fe60c0/ : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":29},{"id":"08def2e9d46f28e6ed84ff8933f94b7d06789b97","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-04 04:00:02.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"28f44abf8641518ca411d716601e5ea2dbac2317","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-04 04:41:55.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/b4b5a60/ : SUCCESS in 4m 06s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/ed5ca33/ : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":29},{"id":"014f20db4bb3079b2ca8442b627d72889527f10f","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-04 18:27:34.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"f3ed7cfff1ea94e3e77343849dcbd5d5a08622f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-04 19:04:35.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/6396cd3/ : SUCCESS in 4m 31s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/4e92c1a/ : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":29},{"id":"f2c7fb3e04604d5ac3621658248343895880b1e6","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-06 22:59:21.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"9ba4e3513d399391a7c16efa0f9846f53440f074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-06 23:53:00.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/9f4e1a0/ : SUCCESS in 4m 14s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/c5d4888/ : FAILURE in 46m 56s","accounts_in_message":[],"_revision_number":29},{"id":"eee71e6e59cb62982a748138d54a3223d3a6b8f3","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-07 02:35:11.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"1aa695e3347427a9cabec01c880be31f80a0e227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-07 03:08:18.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/106daff/ : SUCCESS in 3m 40s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/e7320be/ : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":29},{"id":"f6db86f69b3106122c0a130c21f322b53724ac75","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-07 19:36:33.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"ef35cc981481083acd33ddf66229088b896fc090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-07 20:15:34.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/29/check/openstack-tox-pep8/751653b/ : SUCCESS in 3m 33s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/29/check/barbican-simple-crypto-devstack-tempest/f050c75/ : FAILURE in 32m 57s","accounts_in_message":[],"_revision_number":29},{"id":"990abd23999d73c45a7de2845b14c66de8b25344","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-07 20:28:07.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"d3187b86b078de299cc4eb9410a7367bee65270b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-07 21:09:54.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/79801e9/ : SUCCESS in 3m 21s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/1611a6f/ : FAILURE in 35m 51s","accounts_in_message":[],"_revision_number":30},{"id":"89cae1be3b6742a21bc14fd286825a6a30684ffd","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-07 22:18:48.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"baf13a05c6f9d201f98780171cf45f60c466a8a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-07 22:57:10.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/a0737bb/ : SUCCESS in 4m 16s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/fb708aa/ : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":30},{"id":"6f4d85a4b7b97e3c7be76b05108d9f897e447a11","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-07 23:46:32.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"7e4b829c6c075540d2d73b9dedd47db9a92fb5ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-08 00:25:07.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/4e60586/ : SUCCESS in 3m 46s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/9fb6df2/ : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":30},{"id":"2494fc8a01764139ac1ff156bf7a256b18bc5583","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-10 01:43:01.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"09df48c016ab3d4add2572bc927e32eaa3273736","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 02:20:28.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/d7852df/ : SUCCESS in 3m 42s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/e1da480/ : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":30},{"id":"ded76e49e171abd110a3f155396fb91697f3c454","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-10 21:21:43.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"b4bd01d82fbd48b153936c4286cb2eea6d57b2ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 21:59:23.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/fee32da/ : SUCCESS in 4m 36s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/9bf11d6/ : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":30},{"id":"8073598268122af99f9f7432798b8328c15a5e26","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-11 01:34:23.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"2b3ab705a3aaec4e201b6baf502c978b48a45762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-11 02:06:53.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/5afae93/ : SUCCESS in 3m 54s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/806227a/ : SUCCESS in 28m 27s","accounts_in_message":[],"_revision_number":30},{"id":"daa3bf0eb56acb13bd9f8ceefc36282c8dc2d854","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-11 02:58:18.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"dd6cd677d31bf1f1f2b100a3f14d83131fb766fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-11 03:33:08.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/5f7ebcb/ : SUCCESS in 3m 31s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/be5207a/ : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":30},{"id":"74328963f90591dc68c4f1d7152bec1fbec615a8","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-11 04:22:07.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"e24e2b2096672aa047605f27a3777d996a9f1338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-11 04:57:43.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/30/check/openstack-tox-pep8/c1f0431/ : SUCCESS in 4m 45s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/30/check/barbican-simple-crypto-devstack-tempest/813c71f/ : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":30},{"id":"77b81cb0ed02f38e617a4869b8531f82a302d443","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-11 05:10:37.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"96f585b4d76c12be411a914f6276ca5a589aa225","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-11 05:42:23.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/31/check/openstack-tox-pep8/ccccdca/ : SUCCESS in 3m 16s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/e300266/ : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":31},{"id":"2e391ae702b73f80a433c199530d1ffdfce18442","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-04-13 16:10:39.000000000","message":"Patch Set 31:\n\nrecheck","accounts_in_message":[],"_revision_number":31},{"id":"bd9ea6bb70db4f19817db230a275553427fae697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-13 17:00:40.000000000","message":"Patch Set 31:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/31/check/openstack-tox-pep8/83ddd90/ : SUCCESS in 4m 55s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/ : SUCCESS in 32m 16s","accounts_in_message":[],"_revision_number":31},{"id":"52bfb4defaf6adc40f682cc552b3e40ab7340ced","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-04-13 22:08:47.000000000","message":"Patch Set 31:\n\nJust noticed you have some warnings in here about DuplicateOptError for \u0027fixed_key\u0027. Not sure if that\u0027s usual or if there\u0027s something that needs to be adjusted in conf somewhere:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-cpu.txt.gz#_Apr_13_16_53_19_763924","accounts_in_message":[],"_revision_number":31},{"id":"7bf55362cf8f48f0375242214d4bab0c8632ecae","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-04-13 22:32:48.000000000","message":"Patch Set 31:\n\n(1 comment)\n\nI think this is test_signed_image_invalid_cert_boot_failure:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-api.txt.gz#_Apr_13_16_53_15_887244\n\nwith req-f38e546a-b7c6-4647-aeb6-c79b402e16aa\n\nwhich returns with a 202 from the API and then fails in n-cpu here:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-cpu.txt.gz#_Apr_13_16_53_20_142715\n\nas we expect with no traceback logged. So that looks all good.\n\nThen we have test_signed_image_upload_and_boot:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-api.txt.gz#_Apr_13_16_53_25_885050\n\nwith req-8ac511a3-19aa-49f1-be65-8515bb537c1d\n\nwhich returns with a 202 from the API and cert validation succeeds in n-cpu here:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-cpu.txt.gz#_Apr_13_16_53_29_911382\n\nas we expect. And the instance finishes spawning here, though for some reason the request-id has changed by this point (?):\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-cpu.txt.gz#_Apr_13_16_53_33_502262","accounts_in_message":[],"_revision_number":31},{"id":"65154a9a2d5937a571f0d6fac694a7a41efc7ca8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-04-13 22:40:05.000000000","message":"Patch Set 31:\n\nThere\u0027s one traceback in the n-cpu log related to image signature verification:\n\nhttp://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/5379cd4/logs/screen-n-cpu.txt.gz?level\u003dTRACE#_Apr_13_16_53_42_039848\n\nbut I don\u0027t think it\u0027s related to your blueprint work -- it\u0027s from already existing code signature_utils.get_verifier raising SignatureVerificationError, which will be re-raised and not caught anywhere. Probably needs to be caught alongside of InstanceNotFound, UnexpectedDeletingTaskStateError, and CertificateValidationFailed to avoid logging the traceback.","accounts_in_message":[],"_revision_number":31},{"id":"2b0d4fa52f8884d3cdf20e7a9bdab9806aa4b2b9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-04-16 21:35:58.000000000","message":"Patch Set 31:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":31},{"id":"711a27da1c4180e15d396da2da922bdc04f4f2df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-16 22:15:57.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/31/check/openstack-tox-pep8/6a5250f/ : SUCCESS in 3m 43s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/31/check/barbican-simple-crypto-devstack-tempest/0ff6bcc/ : FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":31},{"id":"0209685e7af5b1e6631396ab5bc59a237c783290","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-16 22:53:04.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"f69c4eefbdcd5aeb70af9b653ae033bd1bd6f981","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-16 23:02:49.000000000","message":"Patch Set 32: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":32},{"id":"492db0282730dc249e939e1fa7a0d040e6cff4e2","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-16 23:03:43.000000000","message":"Patch Set 32:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":32},{"id":"a6c4ec65d243c62d73457a59b5697597c6936a6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-16 23:35:52.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/32/check/openstack-tox-pep8/6727e82/ : SUCCESS in 4m 00s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/32/check/barbican-simple-crypto-devstack-tempest/63c3936/ : SUCCESS in 28m 08s","accounts_in_message":[],"_revision_number":32},{"id":"8d2d5fe110644e6dd6a3c32a41d55dcb61ee3dce","author":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"date":"2018-04-17 20:02:45.000000000","message":"Patch Set 32:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":32},{"id":"0a3e01dfdfa879f7465a369f881dcdad1693943a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 20:34:53.000000000","message":"Patch Set 32:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/32/check/openstack-tox-pep8/aade938/ : SUCCESS in 3m 46s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/32/check/barbican-simple-crypto-devstack-tempest/4f64fb6/ : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":32},{"id":"9c0838e9d5b7c51e6293e6805e3029a855a05a2e","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-05-01 15:45:11.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"38e3eee07408db50941b78484b2faf5c0f548760","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-01 16:28:17.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/33/check/openstack-tox-pep8/988686c/ : SUCCESS in 3m 53s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/33/check/barbican-simple-crypto-devstack-tempest/1fc6716/ : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":33},{"id":"d91707e94d1b7a39253b8659ff8ed7583412a7cd","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-05-22 14:46:56.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"b3a707340c5be267f138dedbb1e269a1cbc9e666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 14:47:13.000000000","message":"Patch Set 33: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":33},{"id":"f2013eb08d34000f893898238c97b8f9525bfed6","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-05-22 14:56:47.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"a33baa5ef343c14155383da698cc691ca4dacd24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 15:39:34.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/33/check/openstack-tox-pep8/cf023fa/ : SUCCESS in 4m 07s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/33/check/barbican-simple-crypto-devstack-tempest/61e4d91/ : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":33},{"id":"3de7f5af24e39b3814f45e049ba14b1de93fcf07","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-05-25 17:56:58.000000000","message":"Patch Set 34: Commit message was updated.","accounts_in_message":[],"_revision_number":34},{"id":"2a4b10040620ddb645297c71456e68a7467a0190","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-05-25 17:59:54.000000000","message":"Patch Set 34:\n\nUpdated commit message to depend on a fix in nova for the \"tempest.lib.exceptions.DeleteErrorException: Resource e5ed7f8b-0ba4-46af-b4e0-517b0a54e671 failed to delete and is in ERROR status\" failure.","accounts_in_message":[],"_revision_number":34},{"id":"9cb96c55c1fe160cee08a057ce9baed0b8fabfaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 18:33:13.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/34/check/openstack-tox-pep8/dce5c61/ : SUCCESS in 3m 57s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/34/check/barbican-simple-crypto-devstack-tempest/a98095f/ : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":34},{"id":"7ceec6ed39de03b927258de0ecb7da4b6bdd73f8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-04 14:21:39.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"812b8a4f3d996740322489250f5059ceba64fa49","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-04 14:23:58.000000000","message":"Patch Set 34: Code-Review-1\n\nUmm, this isn\u0027t actually testing certificate validation in the nova stack:\n\nhttp://logs.openstack.org/10/515210/34/check/barbican-simple-crypto-devstack-tempest/a98095f/logs/screen-n-api.txt.gz#_May_25_18_22_08_055500\n\nglance.enable_certificate_validation \u003d False","accounts_in_message":[],"_revision_number":34},{"id":"2e4140585aa5e04b0c0c34a87c59530ad136c59e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-04 14:26:56.000000000","message":"Patch Set 34:\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"fa4aec602c82c2e899f11610c5daeb8da177f73c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-04 14:36:49.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"81967c22799d4a72f44377c74b9c72132fde712a","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-04 14:40:40.000000000","message":"Patch Set 35: Commit message was updated.","accounts_in_message":[],"_revision_number":35},{"id":"6ef8cd28b8075349e2818792b542c0076abc3432","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-04 14:43:33.000000000","message":"Patch Set 34:\n\n(3 comments)\n\nThank you for the review, @Matt Riedemann.  I have removed the un-needed dependencies from the commit message and will create an additional test as requested.","accounts_in_message":[],"_revision_number":34},{"id":"2941973ceb4b14b9eaef215421e8063cf67116a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-04 15:27:22.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/35/check/openstack-tox-pep8/48a6f2f/ : SUCCESS in 4m 11s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/35/check/barbican-simple-crypto-devstack-tempest/b641809/ : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":35},{"id":"e493114ff4be45c777d56f470d420c29a2d27aa6","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-05 10:38:24.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"5a94c046fd2317345bf57808948d34805bf26ab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 10:42:28.000000000","message":"Patch Set 36: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":36},{"id":"19b55e3c4729a6230953fbd0c4e9ca9f375fd9f8","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-05 10:43:37.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"035ec4901e6585d55944f38fdc20bea65e58829a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 11:24:23.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/36/check/openstack-tox-pep8/106b394/ : FAILURE in 4m 33s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/36/check/barbican-simple-crypto-devstack-tempest/3e44d55/ : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":36},{"id":"722bd3e5ec56e08b5c9076c9f5dbe63ac307e7d9","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-05 16:55:59.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"e86650b316147ece22ba0537a22c4d0e86346a28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 17:44:42.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/37/check/openstack-tox-pep8/824ed94/ : SUCCESS in 3m 36s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/37/check/barbican-simple-crypto-devstack-tempest/a6cc353/ : FAILURE in 41m 55s","accounts_in_message":[],"_revision_number":37},{"id":"11cf5ee79b4ba0c1b9001463055a67217ff7f553","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-05 19:58:58.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"b9ac39992fd4b67f0a8265aeb4773d29a5f62a42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 20:39:52.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/38/check/openstack-tox-pep8/9299f29/ : SUCCESS in 5m 02s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/38/check/barbican-simple-crypto-devstack-tempest/27ebe4b/ : FAILURE in 36m 30s","accounts_in_message":[],"_revision_number":38},{"id":"c85955e5e8408670a7948701a73986633320ae42","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-06 21:48:43.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"a06483514f8d0d6f0b6bfaafdd07df2dc6f56f3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 22:21:35.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/39/check/openstack-tox-pep8/3bc8c97/ : SUCCESS in 4m 22s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/39/check/barbican-simple-crypto-devstack-tempest/c3c928f/ : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":39},{"id":"4278ae76879757b84367f819f659d615a8a7bc59","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-13 13:00:24.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"bbf5c40de22597b99927c91cc79cb9a55f75e6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 14:17:52.000000000","message":"Patch Set 39:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/39/check/openstack-tox-pep8/ca0d6cd/ : SUCCESS in 5m 30s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/39/check/barbican-simple-crypto-devstack-tempest/dd7edf9/ : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":39},{"id":"b726dd5232af67345cea83417eb93333dc8d1b93","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-15 14:02:39.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"ce805aca8d6a9f2cbe741b9733fa0d0d1643271a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-15 15:21:01.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/39/check/openstack-tox-pep8/0b8082d/ : SUCCESS in 4m 07s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/39/check/barbican-simple-crypto-devstack-tempest/c7f110a/ : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":39},{"id":"6b678ad8f6730aeb676b8f34f057df1b3352f1c8","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-15 17:37:47.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"b281a755296325041067ecef67c516891ca6ffa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-15 18:39:43.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/39/check/openstack-tox-pep8/8cc3883/ : SUCCESS in 4m 51s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/39/check/barbican-simple-crypto-devstack-tempest/7b7f049/ : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":39},{"id":"d6d8a2c7a2c7fc13b14038cd6a4f4ffb0229d180","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-18 13:23:56.000000000","message":"Patch Set 40: Commit message was updated.","accounts_in_message":[],"_revision_number":40},{"id":"d555dee5847fe0ce9763d30065c4554791c0d05e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-18 14:12:12.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/40/check/openstack-tox-pep8/127a0f5/ : SUCCESS in 5m 25s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/40/check/barbican-simple-crypto-devstack-tempest/58098db/ : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":40},{"id":"7c7cb7f400adddf4c1a176a9e1e98fefca3d599f","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-21 23:14:52.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"3529de87585beb2b86e55b89859fff65196d475b","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-21 23:16:58.000000000","message":"Uploaded patch set 42: Commit message was updated.","accounts_in_message":[],"_revision_number":42},{"id":"2897a2cc3a29414896fba863bcf6be7a006aee3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 23:52:56.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/42/check/openstack-tox-pep8/4fcea2a/ : SUCCESS in 5m 23s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/42/check/barbican-simple-crypto-devstack-tempest/788c2b0/ : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":42},{"id":"500e6b6dc61f61e23f699725ee85b7aee25e8451","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-22 16:01:09.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"a1125c18d6ef6eb6cb02aed282993c220fa58946","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-22 16:01:43.000000000","message":"Patch Set 44: Patch Set 43 was rebased","accounts_in_message":[],"_revision_number":44},{"id":"be5edd37a716ead8fdfc1aa2a97fa95cf0348830","author":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"date":"2018-06-22 16:20:25.000000000","message":"Patch Set 44:\n\nThe cert validation tests were updated to be run in their own file, instead of alongside the image signing tests, so that the image signing tests could still be run against pike and queens (which don\u0027t have cert validation support).\n\nThe pike barbican tempest plugin tests still pass when run with these updates.  See: http://logs.openstack.org/64/576564/1/check/barbican-simple-crypto-devstack-tempest/2aa0786/job-output.txt.gz#_2018-06-21_23_57_51_747216\n\nThe queens barbican tempest plugin tests also pass when run with these updates.  See: http://logs.openstack.org/19/577319/1/check/barbican-simple-crypto-devstack-tempest/73a2fa9/job-output.txt.gz#_2018-06-21_23_53_47_751712\n\nAll of this patches this was dependent on have now merged.","accounts_in_message":[],"_revision_number":44},{"id":"bec6d8cec24fe9f9ec40d5253d34c27a3739c835","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 16:45:39.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/44/check/openstack-tox-pep8/046acbc/ : SUCCESS in 5m 16s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/44/check/barbican-simple-crypto-devstack-tempest/6953be1/ : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":44},{"id":"1aaacdc48bb3215f5a668a0fbc39d5eadc6e79d5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-06-06 10:13:57.000000000","message":"Patch Set 45: Patch Set 44 was rebased","accounts_in_message":[],"_revision_number":45},{"id":"10c6d3f9aaeb2681b97e032f94355d07c99d7e4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:01:40.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/515210/45/check/openstack-tox-pep8/15c64ca/ : SUCCESS in 6m 09s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/10/515210/45/check/barbican-simple-crypto-devstack-tempest/6cf1f31/ : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":45},{"id":"5180c1214585c811019236519201055335db7935","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-06-06 11:11:06.000000000","message":"Patch Set 45: Code-Review+1\n\nI found this after working on Ib346d383c430d5151d9aafa6e856dd0a7cae8a23 where I noticed tempest.api.compute.servers.test_servers.ServerShowV263Test was racing in some environments and actually attempting to validate image certificates.\n\nMy changes there shouldn\u0027t impact this work but if we wanted to test tempest.api.compute.servers.test_servers.ServerShowV263Test somewhere in Barbican we would need to pass a valid image and certificates through the new configurables I\u0027m introducing.\n\nAlternately we could just remove that test from Tempest and recreate it here.\n\nEither way for now this change LGTM after a rebase and is still passing.","accounts_in_message":[],"_revision_number":45},{"id":"01d1d9f6ff3bb2e6f923faae76de7c752912c3b0","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-08-06 16:33:48.000000000","message":"Patch Set 45: Code-Review+2","accounts_in_message":[],"_revision_number":45},{"id":"49735ea9496ec367e028e90553601994909c966b","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-14 15:47:33.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"645193de361cf1b261a9fb1eae0febdc5061c7a2","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 06:39:58.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"755a3bc7ec5215483fc96050cc829365e562d38c","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-12 08:13:38.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"5f409703c33be88d4bae70c9e519da4bacb36d11","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-10-13 14:39:28.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"06dec125a6f151da88a5d749603c6916ca31c6d2","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-10-13 15:00:03.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"63e48f2b3c03ea2110f84744a338c2ea56340239","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-31 05:41:52.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"f5292bcf18c53aa9fa52f4ba8b674b277498974c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-11-26 09:32:25.000000000","message":"Removed reviewer Lee Yarwood with the following votes:\n\n* Code-Review+1 by Lee Yarwood \u003clyarwood@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":45},{"id":"440ee501bc583337bfc8b17a2d5a6efb5b33378e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-03-13 21:18:57.000000000","message":"Patch Set 46: Patch Set 45 was rebased","accounts_in_message":[],"_revision_number":46},{"id":"be6d673616f5ea16210cbec0f137182f1e245094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:58:13.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5595ba0b22d44d8b5c45a6b21d016e1 : SUCCESS in 3m 22s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/61419a04e3fa459ba2e1576533324621 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":46},{"id":"2535fa0b615e1c02f641074448463cd962a35b51","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-08-25 10:09:58.000000000","message":"Patch Set 46: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":46},{"id":"98e6937824f0ad23a5e08d0d954275147d999ba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 10:10:11.000000000","message":"Patch Set 46: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":46},{"id":"150c38576fb219d65dcaf8f765f8d98db285bc35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:01:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":46},{"id":"c9cf3e968e2471da817e4b8e762013c43251fe34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:01:01.000000000","message":"Patch Set 46: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b6310416ed41f1a9f191072546d4e3 : SUCCESS in 4m 02s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f389917d13b947a4a528e5910071e9db : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":46}],"current_revision_number":46,"current_revision":"f36d46b72499ce6692f3284738e17965f7fe1b99","revisions":{"fc78edf90ff036e2487e768598960336e68493c3":{"kind":"REWORK","_number":1,"created":"2017-10-25 21:10:36.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/1"}}},"commit":{"parents":[{"commit":"078a6c435e3ddb06ab667098135fa8441f5b412a","subject":"Replace deprecated function in tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/078a6c435e3ddb06ab667098135fa8441f5b412a"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate the cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fc78edf90ff036e2487e768598960336e68493c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fc78edf90ff036e2487e768598960336e68493c3"}]},"branch":"refs/heads/master"},"fbfb4f95d9489fe600baae08d33bc5a495897e45":{"kind":"REWORK","_number":2,"created":"2017-10-25 23:05:59.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/2"}}},"commit":{"parents":[{"commit":"078a6c435e3ddb06ab667098135fa8441f5b412a","subject":"Replace deprecated function in tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/078a6c435e3ddb06ab667098135fa8441f5b412a"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 22:59:34.000000000","tz":-240},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate the cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fbfb4f95d9489fe600baae08d33bc5a495897e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fbfb4f95d9489fe600baae08d33bc5a495897e45"}]},"branch":"refs/heads/master"},"17d196705e7ef3a1576eedbf44fd98bb1faf3e24":{"kind":"REWORK","_number":3,"created":"2017-11-30 20:25:26.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/3"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-11-30 20:25:21.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate the cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/17d196705e7ef3a1576eedbf44fd98bb1faf3e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/17d196705e7ef3a1576eedbf44fd98bb1faf3e24"}]},"branch":"refs/heads/master"},"3fc290ef7422f418ea1f353054511d5ff4791620":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2017-12-01 20:49:32.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/4"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-01 20:49:09.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3fc290ef7422f418ea1f353054511d5ff4791620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3fc290ef7422f418ea1f353054511d5ff4791620"}]},"branch":"refs/heads/master"},"0bc19681b98c10feb4cae7cec5dac08906cf523d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2017-12-05 20:52:03.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/5"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2017-12-05 20:51:44.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0bc19681b98c10feb4cae7cec5dac08906cf523d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0bc19681b98c10feb4cae7cec5dac08906cf523d"}]},"branch":"refs/heads/master"},"3aef4630b4c37a9f11a3068f96876d0e5eeb1b16":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2017-12-05 20:54:44.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/6"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2017-12-05 20:53:56.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3aef4630b4c37a9f11a3068f96876d0e5eeb1b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3aef4630b4c37a9f11a3068f96876d0e5eeb1b16"}]},"branch":"refs/heads/master"},"9c4a891847c2885952a17cdcc3153f780af16392":{"kind":"REWORK","_number":7,"created":"2017-12-05 22:24:00.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/7"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-05 22:23:24.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: Iedd3fea0e86648fae364f075915555dcb2c4f199\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9c4a891847c2885952a17cdcc3153f780af16392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9c4a891847c2885952a17cdcc3153f780af16392"}]},"branch":"refs/heads/master"},"aac0e741decd1945ce2791dcf504433b68f708ba":{"kind":"REWORK","_number":8,"created":"2017-12-07 19:26:10.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/8"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-07 19:24:48.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/aac0e741decd1945ce2791dcf504433b68f708ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/aac0e741decd1945ce2791dcf504433b68f708ba"}]},"branch":"refs/heads/master"},"7c98dd917eeffbf7f8bc814af1880b21613b27a3":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2017-12-07 22:33:17.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/9"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-07 22:33:10.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7c98dd917eeffbf7f8bc814af1880b21613b27a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7c98dd917eeffbf7f8bc814af1880b21613b27a3"}]},"branch":"refs/heads/master"},"ac0b4a039b8ad6f2f77dc7f96efd5583379d3d78":{"kind":"REWORK","_number":10,"created":"2017-12-08 18:02:48.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/10"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-08 18:02:36.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ac0b4a039b8ad6f2f77dc7f96efd5583379d3d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ac0b4a039b8ad6f2f77dc7f96efd5583379d3d78"}]},"branch":"refs/heads/master"},"da97c46f4abc225530a58b0077771ee3042a68b0":{"kind":"REWORK","_number":11,"created":"2017-12-08 18:34:04.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/11"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-08 18:33:57.000000000","tz":-300},"subject":"[WIP] Add certificate validation scenario tests","message":"[WIP] Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da97c46f4abc225530a58b0077771ee3042a68b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da97c46f4abc225530a58b0077771ee3042a68b0"}]},"branch":"refs/heads/master"},"009e5a1755bd5290cd8226b5ddc133cb984c3b67":{"kind":"REWORK","_number":12,"created":"2017-12-08 22:30:55.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/12"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-08 22:30:49.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/009e5a1755bd5290cd8226b5ddc133cb984c3b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/009e5a1755bd5290cd8226b5ddc133cb984c3b67"}]},"branch":"refs/heads/master"},"7d9b44d48c3559eeb87af03ba3e53760db7dd692":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2017-12-08 23:51:59.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/13"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-08 23:51:39.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7d9b44d48c3559eeb87af03ba3e53760db7dd692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7d9b44d48c3559eeb87af03ba3e53760db7dd692"}]},"branch":"refs/heads/master"},"64356ecb55822423af43f8a317768371f54e20c2":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2017-12-09 03:55:38.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/14"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-09 03:55:32.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64356ecb55822423af43f8a317768371f54e20c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64356ecb55822423af43f8a317768371f54e20c2"}]},"branch":"refs/heads/master"},"3f0da3dbffda244ed80bd1bfbbf80c75d8e06861":{"kind":"REWORK","_number":15,"created":"2017-12-11 19:51:45.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/15"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-11 19:50:51.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3f0da3dbffda244ed80bd1bfbbf80c75d8e06861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3f0da3dbffda244ed80bd1bfbbf80c75d8e06861"}]},"branch":"refs/heads/master"},"534ed0ab6b067d26ee8c6d95699553b5fe802f13":{"kind":"REWORK","_number":16,"created":"2017-12-11 21:55:45.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/16"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-11 21:55:20.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/534ed0ab6b067d26ee8c6d95699553b5fe802f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/534ed0ab6b067d26ee8c6d95699553b5fe802f13"}]},"branch":"refs/heads/master"},"c306e21080317149c524e2d89a301035a0078fe0":{"kind":"REWORK","_number":17,"created":"2017-12-12 22:35:33.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/17"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-12 22:35:17.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nDepends-On: I40268822c6a2c6ef245f03863c2c3379da50acd7\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c306e21080317149c524e2d89a301035a0078fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c306e21080317149c524e2d89a301035a0078fe0"}]},"branch":"refs/heads/master"},"349df0ce074a1db05349118aa343b921834a0b9e":{"kind":"NO_CHANGE","_number":18,"created":"2017-12-13 02:04:16.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/18"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-13 02:04:00.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nDepends-On: I40268822c6a2c6ef245f03863c2c3379da50acd7\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/349df0ce074a1db05349118aa343b921834a0b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/349df0ce074a1db05349118aa343b921834a0b9e"}]},"branch":"refs/heads/master"},"32d5cd2225142007c7bf0c1d5c140c2991f434b6":{"kind":"REWORK","_number":19,"created":"2017-12-13 15:55:44.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/19"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-13 15:55:01.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature, which uses\nNova API microversion 2.57.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nDepends-On: I40268822c6a2c6ef245f03863c2c3379da50acd7\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/32d5cd2225142007c7bf0c1d5c140c2991f434b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/32d5cd2225142007c7bf0c1d5c140c2991f434b6"}]},"branch":"refs/heads/master"},"c98f924ea8a9c8d05115009dfcbb54e5063f585e":{"kind":"REWORK","_number":20,"created":"2017-12-13 21:51:47.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/20"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-13 21:51:35.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nDepends-On: I40268822c6a2c6ef245f03863c2c3379da50acd7\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c98f924ea8a9c8d05115009dfcbb54e5063f585e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c98f924ea8a9c8d05115009dfcbb54e5063f585e"}]},"branch":"refs/heads/master"},"3dc4a3b2bd52f45a6bf7c5eac8b80f5f645b91f4":{"kind":"REWORK","_number":21,"created":"2017-12-14 16:38:38.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/21"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-14 16:37:39.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3dc4a3b2bd52f45a6bf7c5eac8b80f5f645b91f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3dc4a3b2bd52f45a6bf7c5eac8b80f5f645b91f4"}]},"branch":"refs/heads/master"},"1fbeb0a077d22935a0e12f19522791aa51292e58":{"kind":"REWORK","_number":22,"created":"2017-12-16 00:27:43.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/22"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-16 00:27:20.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1fbeb0a077d22935a0e12f19522791aa51292e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1fbeb0a077d22935a0e12f19522791aa51292e58"}]},"branch":"refs/heads/master"},"2376b3d2b5b3e57e738753b7930211145fe85802":{"kind":"REWORK","_number":23,"created":"2017-12-16 02:40:10.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/23"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-16 02:39:57.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2376b3d2b5b3e57e738753b7930211145fe85802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2376b3d2b5b3e57e738753b7930211145fe85802"}]},"branch":"refs/heads/master"},"0ee9576230873b404033d7b081556b34e69ed334":{"kind":"REWORK","_number":24,"created":"2017-12-16 20:38:13.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/24"}}},"commit":{"parents":[{"commit":"da8979b10fa6733bac8fb43dce35a14f8af49301","subject":"Add .zuul.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da8979b10fa6733bac8fb43dce35a14f8af49301"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-12-16 20:37:41.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0ee9576230873b404033d7b081556b34e69ed334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0ee9576230873b404033d7b081556b34e69ed334"}]},"branch":"refs/heads/master"},"3ac6210186b67a4812936227582c7fa27527889a":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2018-01-22 15:22:47.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/25"}}},"commit":{"parents":[{"commit":"f3dcdd472a9bb78bc48c18473ed646924a7184fd","subject":"Add connection test to help diagnose failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f3dcdd472a9bb78bc48c18473ed646924a7184fd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-01-22 15:22:47.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3ac6210186b67a4812936227582c7fa27527889a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3ac6210186b67a4812936227582c7fa27527889a"}]},"branch":"refs/heads/master"},"97020e42c796bd07a82b8b8668c7b9d0c4973fbf":{"kind":"REWORK","_number":26,"created":"2018-01-22 16:12:47.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/26"}}},"commit":{"parents":[{"commit":"f3dcdd472a9bb78bc48c18473ed646924a7184fd","subject":"Add connection test to help diagnose failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f3dcdd472a9bb78bc48c18473ed646924a7184fd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-01-22 16:12:25.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/97020e42c796bd07a82b8b8668c7b9d0c4973fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/97020e42c796bd07a82b8b8668c7b9d0c4973fbf"}]},"branch":"refs/heads/master"},"532e63788935b5c7d04789499e8806779744ec73":{"kind":"REWORK","_number":27,"created":"2018-01-23 15:12:34.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/27"}}},"commit":{"parents":[{"commit":"5db05b85a2e90150b6adceb21cd4d03b896eb124","subject":"Merge \"Fixed _log_console_output call on failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5db05b85a2e90150b6adceb21cd4d03b896eb124"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-01-23 15:12:27.000000000","tz":-300},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/532e63788935b5c7d04789499e8806779744ec73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/532e63788935b5c7d04789499e8806779744ec73"}]},"branch":"refs/heads/master"},"eb022aea3d190d762972418e893b1f8739f8ccd1":{"kind":"REWORK","_number":28,"created":"2018-03-22 19:05:29.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/28"}}},"commit":{"parents":[{"commit":"78f3327364345164a52ac6bbf0c7df0fa1ae8951","subject":"Verify instance snapshots when using signed images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/78f3327364345164a52ac6bbf0c7df0fa1ae8951"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-03-22 19:05:22.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/eb022aea3d190d762972418e893b1f8739f8ccd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/eb022aea3d190d762972418e893b1f8739f8ccd1"}]},"branch":"refs/heads/master"},"c079b8e3015e49934d41c69ffee3356a87ba46ed":{"kind":"REWORK","_number":29,"created":"2018-03-23 13:35:07.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/29"}}},"commit":{"parents":[{"commit":"78f3327364345164a52ac6bbf0c7df0fa1ae8951","subject":"Verify instance snapshots when using signed images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/78f3327364345164a52ac6bbf0c7df0fa1ae8951"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-03-23 13:35:05.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c079b8e3015e49934d41c69ffee3356a87ba46ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c079b8e3015e49934d41c69ffee3356a87ba46ed"}]},"branch":"refs/heads/master"},"5ae360624f2593aad869258a5bdf8fb39feb1a67":{"kind":"REWORK","_number":30,"created":"2018-04-07 20:28:07.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/30"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-04-07 20:27:38.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ae360624f2593aad869258a5bdf8fb39feb1a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ae360624f2593aad869258a5bdf8fb39feb1a67"}]},"branch":"refs/heads/master"},"bfa488e3e90dd0080aee502b8f58da4fd83e9404":{"kind":"REWORK","_number":31,"created":"2018-04-11 05:10:37.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/31"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-04-11 05:10:25.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/bfa488e3e90dd0080aee502b8f58da4fd83e9404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/bfa488e3e90dd0080aee502b8f58da4fd83e9404"}]},"branch":"refs/heads/master"},"5a96f367e7783bb2d6f64ee499f9521879e27ab3":{"kind":"REWORK","_number":32,"created":"2018-04-16 22:53:04.000000000","uploader":{"_account_id":24711,"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","username":"jackie-truong"},"ref":"refs/changes/10/515210/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/32"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2018-04-16 22:52:58.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5a96f367e7783bb2d6f64ee499f9521879e27ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5a96f367e7783bb2d6f64ee499f9521879e27ab3"}]},"branch":"refs/heads/master"},"cba1da6ef4720b379296bb532a06de2369d6daf2":{"kind":"REWORK","_number":33,"created":"2018-05-01 15:45:11.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/33"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-05-01 15:45:05.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: I0ae2dbf66241207a425bf7d0fc02a4d2e2dea409\nDepends-On: I48f62d36a871a95601ebf277a2f39a41a873daa8\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/cba1da6ef4720b379296bb532a06de2369d6daf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/cba1da6ef4720b379296bb532a06de2369d6daf2"}]},"branch":"refs/heads/master"},"4d35f48f8525b3cde9029e02cf5a287d92050c3e":{"kind":"NO_CODE_CHANGE","_number":34,"created":"2018-05-25 17:56:58.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/34"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-05-25 17:56:33.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/479949/\nDepends-On: https://review.openstack.org/#/c/526485/\nDepends-On: https://review.openstack.org/#/c/569062/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4d35f48f8525b3cde9029e02cf5a287d92050c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4d35f48f8525b3cde9029e02cf5a287d92050c3e"}]},"branch":"refs/heads/master"},"45a81deedb582b683e2d4e675004148b6659872e":{"kind":"NO_CODE_CHANGE","_number":35,"created":"2018-06-04 14:40:40.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/35"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-04 14:40:36.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/45a81deedb582b683e2d4e675004148b6659872e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/45a81deedb582b683e2d4e675004148b6659872e"}]},"branch":"refs/heads/master"},"c45438e24c21a1cbd8a4599f1326e1d555f21071":{"kind":"REWORK","_number":36,"created":"2018-06-05 10:38:24.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/36"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-05 10:34:53.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c45438e24c21a1cbd8a4599f1326e1d555f21071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c45438e24c21a1cbd8a4599f1326e1d555f21071"}]},"branch":"refs/heads/master"},"3665c787c018e80881ff96f1bd4e5d3a6ebced62":{"kind":"REWORK","_number":37,"created":"2018-06-05 16:55:59.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/37"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-05 16:55:40.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3665c787c018e80881ff96f1bd4e5d3a6ebced62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3665c787c018e80881ff96f1bd4e5d3a6ebced62"}]},"branch":"refs/heads/master"},"6007b7e49f5906edf059c5c872ce01a93e0d6f5e":{"kind":"REWORK","_number":38,"created":"2018-06-05 19:58:58.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/38"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-05 19:58:51.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6007b7e49f5906edf059c5c872ce01a93e0d6f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6007b7e49f5906edf059c5c872ce01a93e0d6f5e"}]},"branch":"refs/heads/master"},"4a65d6de37e5bd5d3fe0b21bb1016968d3338c63":{"kind":"REWORK","_number":39,"created":"2018-06-06 21:48:43.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/39"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-06 21:48:32.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4a65d6de37e5bd5d3fe0b21bb1016968d3338c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4a65d6de37e5bd5d3fe0b21bb1016968d3338c63"}]},"branch":"refs/heads/master"},"a0923f24c8b4ccf29917d2466cabb03a073b423b":{"kind":"NO_CODE_CHANGE","_number":40,"created":"2018-06-18 13:23:56.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/10/515210/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/40"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-06-18 13:23:52.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/576136/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a0923f24c8b4ccf29917d2466cabb03a073b423b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a0923f24c8b4ccf29917d2466cabb03a073b423b"}]},"branch":"refs/heads/master"},"b59bc39a2285b7c1ffa58360e899c7f8d27281b6":{"kind":"REWORK","_number":41,"created":"2018-06-21 23:14:52.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/41"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-21 23:14:31.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch updates the existing image signing scenario tests to\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nDepends-On: https://review.openstack.org/#/c/526485/\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b59bc39a2285b7c1ffa58360e899c7f8d27281b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b59bc39a2285b7c1ffa58360e899c7f8d27281b6"}]},"branch":"refs/heads/master"},"60e05e626f77216d0feb86ce39b548ae36fa8d87":{"kind":"NO_CODE_CHANGE","_number":42,"created":"2018-06-21 23:16:58.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/42"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-21 23:16:09.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch adds certificate validation scenario tests\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/60e05e626f77216d0feb86ce39b548ae36fa8d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/60e05e626f77216d0feb86ce39b548ae36fa8d87"}]},"branch":"refs/heads/master"},"47ac44311b697a44d7af8996fbd208c9b0b1cac2":{"kind":"REWORK","_number":43,"created":"2018-06-22 16:01:09.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/43"}}},"commit":{"parents":[{"commit":"5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd","subject":"Use addCleanup to detach volume instead of explicit detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5ddcc7eefaefb4ef178081ec4838e8f450c6c1dd"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-21 23:25:33.000000000","tz":-240},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch adds certificate validation scenario tests\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/47ac44311b697a44d7af8996fbd208c9b0b1cac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/47ac44311b697a44d7af8996fbd208c9b0b1cac2"}]},"branch":"refs/heads/master"},"8953606a38154e8966ac3325f61aacfb0a851a48":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2018-06-22 16:01:43.000000000","uploader":{"_account_id":7012,"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","username":"brianna-poulos"},"ref":"refs/changes/10/515210/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/44"}}},"commit":{"parents":[{"commit":"c9f0b5d46b17bc55182489150bd6cc6a0399d204","subject":"Added support of volume api_v3 client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c9f0b5d46b17bc55182489150bd6cc6a0399d204"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Brianna Poulos","email":"Brianna.Poulos@jhuapl.edu","date":"2018-06-22 16:01:43.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch adds certificate validation scenario tests\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8953606a38154e8966ac3325f61aacfb0a851a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8953606a38154e8966ac3325f61aacfb0a851a48"}]},"branch":"refs/heads/master"},"439e1451dba3f97657e208751aa164580d51eeeb":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2019-06-06 10:13:57.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/10/515210/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/45"}}},"commit":{"parents":[{"commit":"375a7d69bd1c02979a2e870aebfe8382cc5d17e7","subject":"Merge \"Add option to toggle validation of signed image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/375a7d69bd1c02979a2e870aebfe8382cc5d17e7"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-06-06 10:13:57.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch adds certificate validation scenario tests\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/439e1451dba3f97657e208751aa164580d51eeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/439e1451dba3f97657e208751aa164580d51eeeb"}]},"branch":"refs/heads/master"},"f36d46b72499ce6692f3284738e17965f7fe1b99":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2020-03-13 21:18:57.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/10/515210/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/10/515210/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/10/515210/46"}}},"commit":{"parents":[{"commit":"d1585db02955383c995aea4ef525e985aa82a2ed","subject":"Merge \"Update the old doc links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d1585db02955383c995aea4ef525e985aa82a2ed"}]}],"author":{"name":"Jackie Truong","email":"jacklyn.truong@jhuapl.edu","date":"2017-10-25 21:06:12.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-03-13 21:18:57.000000000","tz":0},"subject":"Add certificate validation scenario tests","message":"Add certificate validation scenario tests\n\nThis patch adds certificate validation scenario tests\nincorporate cursive\u0027s new certificate validation feature.\n\nChange-Id: Ica1d32ca66d6f89692e9cc90e415bf569332f58d\nImplements: blueprint nova-validate-certificates\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f36d46b72499ce6692f3284738e17965f7fe1b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f36d46b72499ce6692f3284738e17965f7fe1b99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
