)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~633307","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~If0be3453cf9a9f2872733c76fc1f791339450d40","project":"openstack/barbican-tempest-plugin","branch":"master","hashtags":[],"change_id":"If0be3453cf9a9f2872733c76fc1f791339450d40","subject":"Fix version in test-requirements.txt","status":"ABANDONED","created":"2019-01-26 03:37:05.000000000","updated":"2021-02-16 13:23:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"588286cc5533749c4a1757a72be35dd4bce3128c","_number":633307,"virtual_id_number":633307,"owner":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":-1,"date":"2019-01-26 04:24:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"all":[{"value":-1,"date":"2019-06-28 16:21:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":-1,"date":"2021-02-15 21:26:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-28 16:17:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-26 04:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-28 16:17:10.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-28 16:21:52.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2021-02-15 21:26:55.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"b789f873879072f43f4627f97f22c1a28f217ada","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2019-01-26 03:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0ff0823da1d4cc0c86406267488a7eee0336a72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 04:24:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/633307/1/check/openstack-tox-pep8/831b998/ : FAILURE in 6m 15s\n- requirements-check http://logs.openstack.org/07/633307/1/check/requirements-check/ccf52d2/ : SUCCESS in 4m 26s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/07/633307/1/check/barbican-simple-crypto-devstack-tempest/a1f2489/ : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":1},{"id":"008c33fbc93caea5d54b7f870bb6ec28e7f29d98","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-28 16:17:10.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"e9011a585065640400ce03bf6f12f17efdb04b70","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-06-28 16:21:52.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"5729453895d289618e2dbce563362cb6d257b0eb","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-15 21:26:55.000000000","message":"Patch Set 1: Code-Review-1\n\nThis old change can be dropped at this point","accounts_in_message":[],"_revision_number":1},{"id":"588286cc5533749c4a1757a72be35dd4bce3128c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-02-16 13:23:29.000000000","message":"Abandoned\n\nThis patch is no longer needed.  Hacking verison was already updated in https://opendev.org/openstack/barbican-tempest-plugin/commit/a318f6d77c0b354b4aee2fe9d7651d595adcf38e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03292eddcd40941852bd9fa34b101bf5fdb1f42b","revisions":{"03292eddcd40941852bd9fa34b101bf5fdb1f42b":{"kind":"REWORK","_number":1,"created":"2019-01-26 03:37:05.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/07/633307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/633307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/633307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/633307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/633307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/633307/1"}}},"commit":{"parents":[{"commit":"123dd7d4162c39a1cb4b4c1b09a5dceaee127eb6","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/123dd7d4162c39a1cb4b4c1b09a5dceaee127eb6"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-26 03:35:39.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-26 03:36:52.000000000","tz":480},"subject":"Fix version in test-requirements.txt","message":"Fix version in test-requirements.txt\n\nChange-Id: If0be3453cf9a9f2872733c76fc1f791339450d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/03292eddcd40941852bd9fa34b101bf5fdb1f42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/03292eddcd40941852bd9fa34b101bf5fdb1f42b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
