)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~710759","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I92098c7d994dc7d85643da29380866a805f4e869","project":"openstack/barbican-tempest-plugin","branch":"master","hashtags":[],"change_id":"I92098c7d994dc7d85643da29380866a805f4e869","subject":"Update attach logic in VolumeEncryptionTest","status":"NEW","created":"2020-03-02 12:32:09.000000000","updated":"2021-03-02 14:54:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"319897dbf51afd9dea5ed59fb75efd0f5a9d5b20","_number":710759,"virtual_id_number":710759,"owner":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-02 14:54:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 12:32:09.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2020-03-02 12:32:42.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2020-03-13 21:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 14:00:40.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"4c69d037870a672e870c04f1b478425e6af387e4","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-03-02 12:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8c03abb1eb8028b4d061f8efccf11fed295f512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 13:33:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c4382ecf744800ba967861880fe80b : SUCCESS in 5m 21s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/14cf742b82804404851ef8947ff0fe45 : FAILURE in 55m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8bbfb29b9785ad88ff8ef4aec0ee6d35414da7cb","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-03-13 21:06:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3dc07fd102ebbd3e2ab5d500917ba1c474cbcc2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:53:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b402e7c1e883494db25ee1874f077048 : SUCCESS in 2m 48s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/6598de3226714f57a38c5ea79224b9fb : FAILURE in 44m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7c0ebb644b605b7bd197d0ef7046549a85528eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-02 14:00:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"319897dbf51afd9dea5ed59fb75efd0f5a9d5b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:54:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9aeb2434aa149a9b028bd5164af88c6 : SUCCESS in 4m 32s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3cd41e7579f84184b08ae1ea2ebb3026 : FAILURE in 40m 26s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/6137b709ac7a4aaf9abadf11067c2ccf : FAILURE in 41m 09s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/04f4f6f7a94645d383777b74450408b7 : FAILURE in 51m 46s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/c5306b60ab7a44b4a2b8a5662e46b07d : FAILURE in 39m 00s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/0dabd3f8428e4e2aa6d71d050c948784 : FAILURE in 51m 12s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b038ed4ce4f1446f8ca7f3dcfce3cad4 : FAILURE in 44m 03s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/37f93109329d405e9e9e11b1b9e5951d : FAILURE in 38m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aa6478442b4bc41acaae0ba34c14ba48c8bd896b","revisions":{"764411c0c35f4e17be3242bbb70fce45acf0ddaf":{"kind":"REWORK","_number":1,"created":"2020-03-02 12:32:09.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/59/710759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/710759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/1"}}},"commit":{"parents":[{"commit":"a6cec8acfd00a5975721b8ea066b90cca55e9920","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a6cec8acfd00a5975721b8ea066b90cca55e9920"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2018-11-04 18:34:58.000000000","tz":120},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-03-02 12:30:14.000000000","tz":120},"subject":"Update attach logic in VolumeEncryptionTest","message":"Update attach logic in VolumeEncryptionTest\n\nPick the same logic as volume attach tests have, when after\nvolume attach server is rebooted to make sure guest OS\nrecognize new device.\n\nChange-Id: I92098c7d994dc7d85643da29380866a805f4e869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/764411c0c35f4e17be3242bbb70fce45acf0ddaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/764411c0c35f4e17be3242bbb70fce45acf0ddaf"}]},"branch":"refs/heads/master"},"62e61eb4ed045a02e8d7ae4c455bf4f7b7d7f96f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-13 21:06:43.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/59/710759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/710759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/2"}}},"commit":{"parents":[{"commit":"d1585db02955383c995aea4ef525e985aa82a2ed","subject":"Merge \"Update the old doc links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d1585db02955383c995aea4ef525e985aa82a2ed"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2018-11-04 18:34:58.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-03-13 21:06:43.000000000","tz":0},"subject":"Update attach logic in VolumeEncryptionTest","message":"Update attach logic in VolumeEncryptionTest\n\nPick the same logic as volume attach tests have, when after\nvolume attach server is rebooted to make sure guest OS\nrecognize new device.\n\nChange-Id: I92098c7d994dc7d85643da29380866a805f4e869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62e61eb4ed045a02e8d7ae4c455bf4f7b7d7f96f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62e61eb4ed045a02e8d7ae4c455bf4f7b7d7f96f"}]},"branch":"refs/heads/master"},"aa6478442b4bc41acaae0ba34c14ba48c8bd896b":{"kind":"REWORK","_number":3,"created":"2021-03-02 14:00:40.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/59/710759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/710759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/710759/3"}}},"commit":{"parents":[{"commit":"ad7f742ea9edd5508123b321d46ec2f4e1af8271","subject":"Merge \"Fix test_signed_image_upload_boot_failure skip method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ad7f742ea9edd5508123b321d46ec2f4e1af8271"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2018-11-04 18:34:58.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-03-02 14:00:32.000000000","tz":60},"subject":"Update attach logic in VolumeEncryptionTest","message":"Update attach logic in VolumeEncryptionTest\n\nPick the same logic as volume attach tests have, when after\nvolume attach server is rebooted to make sure guest OS\nrecognize new device.\n\nChange-Id: I92098c7d994dc7d85643da29380866a805f4e869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/aa6478442b4bc41acaae0ba34c14ba48c8bd896b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/aa6478442b4bc41acaae0ba34c14ba48c8bd896b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
