)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~715514","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I2d2cd5d660ef937bfe8445bae7cf6b279eb80b0f","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"python-requires","hashtags":[],"change_id":"I2d2cd5d660ef937bfe8445bae7cf6b279eb80b0f","subject":"Fix setup.cfg keywords","status":"MERGED","created":"2020-03-27 18:44:02.000000000","updated":"2020-09-25 15:04:35.000000000","submitted":"2020-09-25 15:04:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715514-1601046275199-60701959","meta_rev_id":"af9875ad80083013989762e4fad138db4f374f22","_number":715514,"virtual_id_number":715514,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 15:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-09-25 09:47:40.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2020-09-25 09:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2020-09-25 14:16:11.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 14:34:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-25 16:26:21.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-09-24 13:56:25.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-09-25 14:16:11.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-25 15:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"347657b56ff3955f01c8f8be17a255751c127c06","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-27 18:44:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c52d858c9e70a42d5a8b658045d667bf37b1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 19:30:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4fd8e3cca8894d89bbeee8acfb9b6a0d : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d45d307d414a15acce6a88e5e2acd8 : SUCCESS in 2m 52s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/0d9539340612484d8b8563c64a8b4f3c : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e53f516317344f67342ba17c3cf21601cc858496","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 14:34:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"50b87cc9297f2c2627b1344b869ce7255412ad90","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-25 16:26:21.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49d45a1cc46176906c40ce6525ba9d2ad582accb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 17:17:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/03005d05a9064c8d869e9ac780b3d6b3 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d75f82693a4870be346c51494ee99f : SUCCESS in 3m 51s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/e0ecdfe8089d49bebd7f091a518b5da5 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b38f4d26e2ae67aff0a4f6466002fbf50c611df5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-08-18 13:23:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64ecd6a522ed13dc2e88127338d55a02937be9a5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-01 13:31:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e0851711554767f559c3bd7f64da8eca17df961","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-22 13:31:28.000000000","message":"Removed Code-Review+2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"bacf2ae881538eb2f026b0cb3b9e7032f98f2010","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-22 13:40:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"562848b4ea6dd8e0db071c2e47ccf77877c4badb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 14:21:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db22c975b4cc4d96af074e0bf58b9654 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd2141ffc05479594f572efa528ad01 : SUCCESS in 3m 06s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/487146749cbe457ebc85d796a988c2c3 : SUCCESS in 32m 53s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/1d204cc41e2c4aeea6ec18d63dfdbd1b : SUCCESS in 31m 26s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/9b1274f327cd4077a9671640f8095392 : SUCCESS in 34m 04s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/f814e125f56b47488025db8c35e0b47d : SUCCESS in 31m 44s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5a70538f89f942d4bef3f5e1ad998dfc : SUCCESS in 34m 03s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/402fb6ecf56e47969aeb1f612a842839 : SUCCESS in 37m 29s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ea6b031fafe7468d8db333a2373d70b5 : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":2},{"id":"cf6d4597544515ba7a57177a1de34d305e08e194","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-23 08:06:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c33744ce3c702ecca6e951ddb7752ebdf4433a43","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-24 13:56:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0ef309059eefbab45309a14a479bbbc5eaee22a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:56:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea2e196fffb7b6c0457dac8b26118c04840455c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:01:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8cef42e702447be80ea3c3711e4c33c : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba00c2b4d6f429bacea979f45ee24b5 : SUCCESS in 3m 50s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/72d53d966f834e718af091d1f9cd17f1 : FAILURE in 52m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3a21d0449c24e87c59ad6ead813b92e255427dc9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-24 17:21:36.000000000","message":"Patch Set 2:\n\nThe gate failure should be fixed when https://review.opendev.org/#/c/753699/ lands","accounts_in_message":[],"_revision_number":2},{"id":"8b08dfdb2c7674e540288290318453c357fc44ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:51:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d87a0a3eb44f48d689be9d07f2b8d23d : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9762104cd55d48b3a470a072d1bf6405 : SUCCESS in 3m 46s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/735a497868f34c8fa5474b8ecf6e59a4 : FAILURE in 32m 52s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/12d8e64f4c94468d956e5178be140664 : SUCCESS in 48m 51s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/023abd48bdf9467ca5073d7037b4d1ea : SUCCESS in 42m 46s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/fd399586586d415ebd243037a76b4360 : SUCCESS in 40m 58s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/47d590b28ef945fcb4bbc9492b4bc37a : FAILURE in 42m 16s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/89ef3458105d4fd3a37ba40a55bc3177 : FAILURE in 35m 11s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8e69662e267b46d7a63cdee96d02a3bb : FAILURE in 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fdda7b45cc5b2e5124f8b59bc9cbd843e4db4366","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-25 07:41:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"61c35409361838b6d80c40bafb7fdc5691be20f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:59:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b853c199cd543ff8aabe697e19c3614 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a635a95107f402ca120119327f17974 : SUCCESS in 3m 37s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3b10d385995846eb82541ddaa1229ef0 : FAILURE in 42m 36s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/d7121feb0b744007ba6a31ae41660ac2 : SUCCESS in 31m 24s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/7c7925031f1e42f6a4be9c9701358b2a : SUCCESS in 46m 18s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/52c80e5096614b0f8d0395d34f48f0ad : SUCCESS in 31m 34s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b818950743b04869abba74190f98e35d : FAILURE in 38m 08s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f06983fc1adb486ca375fa49dc27a705 : FAILURE in 38m 18s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a86a0b3a363a4e998338effb84ffeb1a : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0cede7eeefdb250226ae59d895e0b557a7025534","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-25 09:47:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d489c98803f3c587d660acbe344c7a15f9f44db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 11:07:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/588ecd5139b8443fabf95ea01c355351 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd43fe4782584d4e996d1aeb8e3201d9 : SUCCESS in 3m 59s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2156b923aa204334a87d742a8476d056 : SUCCESS in 42m 32s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/a08eb93cd0d647ebb53b2cee9ceb5479 : SUCCESS in 42m 20s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/7fe5e0f892924084aa296c2dcc1448b1 : SUCCESS in 46m 28s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/1e371720fe444f0ca115bbe60e425918 : SUCCESS in 52m 21s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/e464774f45294853b78819afef2abe3c : SUCCESS in 45m 24s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/005add76a8c9460dbda25888d26854c4 : SUCCESS in 48m 01s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/17e8d74a74b44d39babcb94c35e4e661 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":3},{"id":"cf1d90215592cdcccf8fbeeddaadcfaa99f31dce","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-25 14:16:11.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"faed993acde3fcad844c64952c89226f719c1875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:16:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a23050edb9ffcb6c9c628e0d6f9684f5a5f00fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:04:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/80f9e77c0d624d68b7afb1b2f146dc35 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6fde11eb99d4f349b9f06478c162a14 : SUCCESS in 3m 11s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1561dd5dec6544469903062ea48fe2db : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":3},{"id":"5ad9c669e3941584809ef843985f0814381ac2d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:04:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"826e26fb593398635c317df40161e15c54ec567f","revisions":{"64d95fe2a91a2e548139bb414fbef953a4623a6e":{"kind":"REWORK","_number":1,"created":"2020-03-27 18:44:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/715514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/715514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/1"}}},"commit":{"parents":[{"commit":"57565a5104739e1c84ba0b3523e8f998e344d279","subject":"Merge \"Remove use of deprecated config options for volume API selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/57565a5104739e1c84ba0b3523e8f998e344d279"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:43:38.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:44:01.000000000","tz":-300},"subject":"Replace old requires-python with python-requires","message":"Replace old requires-python with python-requires\n\nrequires-python is old value and does not work, python-requires\nis right metadata.\n\n\u0027python-requires\u0027 is introduced from PEP440 and requires-python was old value\nin PEP345.\n- https://packaging.python.org/guides/distributing-packages-using-setuptools/#python-requires\n- https://www.python.org/dev/peps/pep-0345/#requires-python\n\nChange-Id: I2d2cd5d660ef937bfe8445bae7cf6b279eb80b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64d95fe2a91a2e548139bb414fbef953a4623a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64d95fe2a91a2e548139bb414fbef953a4623a6e"}]},"branch":"refs/heads/master"},"b5cacf3fcc4570429910c98ce8ffd4e270fd1031":{"kind":"REWORK","_number":2,"created":"2020-09-22 13:40:47.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/14/715514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/715514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/2"}}},"commit":{"parents":[{"commit":"57565a5104739e1c84ba0b3523e8f998e344d279","subject":"Merge \"Remove use of deprecated config options for volume API selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/57565a5104739e1c84ba0b3523e8f998e344d279"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:43:38.000000000","tz":-300},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-09-22 13:40:27.000000000","tz":0},"subject":"Fix setup.cfg keywords","message":"Fix setup.cfg keywords\n\nauthor-email, home-page are aliases, replacing them for their final value\n\ndescription-file -\u003e long_description \u003d file:\n\nrequires-python is old value and does not work, python-requires\nis right metadata.\n\n\u0027python-requires\u0027 is introduced from PEP440 and requires-python was old value\nin PEP345.\n- https://packaging.python.org/guides/distributing-packages-using-setuptools/#python-requires\n- https://www.python.org/dev/peps/pep-0345/#requires-python\n\nChange-Id: I2d2cd5d660ef937bfe8445bae7cf6b279eb80b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b5cacf3fcc4570429910c98ce8ffd4e270fd1031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b5cacf3fcc4570429910c98ce8ffd4e270fd1031"}]},"branch":"refs/heads/master"},"826e26fb593398635c317df40161e15c54ec567f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-25 09:47:40.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/14/715514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/715514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/715514/3"}}},"commit":{"parents":[{"commit":"9fcc04ff76b6196532bad198837c2e17fbb5d2cb","subject":"Merge \"Ceph Multibackend changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fcc04ff76b6196532bad198837c2e17fbb5d2cb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:43:38.000000000","tz":-300},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-09-25 09:47:40.000000000","tz":0},"subject":"Fix setup.cfg keywords","message":"Fix setup.cfg keywords\n\nauthor-email, home-page are aliases, replacing them for their final value\n\ndescription-file -\u003e long_description \u003d file:\n\nrequires-python is old value and does not work, python-requires\nis right metadata.\n\n\u0027python-requires\u0027 is introduced from PEP440 and requires-python was old value\nin PEP345.\n- https://packaging.python.org/guides/distributing-packages-using-setuptools/#python-requires\n- https://www.python.org/dev/peps/pep-0345/#requires-python\n\nChange-Id: I2d2cd5d660ef937bfe8445bae7cf6b279eb80b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/826e26fb593398635c317df40161e15c54ec567f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/826e26fb593398635c317df40161e15c54ec567f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
