)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~751559","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ib6290ed1ad5779e4e6fed449f3335dbdada42f0d","project":"openstack/barbican-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ib6290ed1ad5779e4e6fed449f3335dbdada42f0d","subject":"Ceph Multibackend changes","status":"MERGED","created":"2020-09-12 01:01:12.000000000","updated":"2020-09-23 17:05:04.000000000","submitted":"2020-09-23 17:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751559-1600880704381-193fd9bf","meta_rev_id":"c9b1d78413ee2445eef106ab0f6026a57f6d9019","_number":751559,"virtual_id_number":751559,"owner":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 17:05:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2020-09-23 16:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2020-09-23 15:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-09-23 16:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"date":"2020-09-23 15:12:24.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2020-09-23 16:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 15:12:24.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-23 15:15:35.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-23 16:01:40.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-09-23 16:11:07.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-09-23 17:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b7c0774c93da1528ce030507a94f493699ab774","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-09-12 01:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d8452a50a55079371666a6fb213b25bfacac6d8","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-09-12 01:02:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2fb30d274fec2b72a1a52d078bf62a60506e1a30","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-09-12 01:04:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"caae1241ed1b65c4e21e97c9c347d5e268d3aa89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 01:56:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538e4f2401e04d05b757b834a0102260 : SUCCESS in 3m 21s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ba2bc16c6c924e4d92c5025c38e4cd9c : SUCCESS in 42m 20s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/1dd49093c1a14399a54309ba1f891754 : SUCCESS in 44m 23s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/e45203ae173845db8f9972e965369a91 : SUCCESS in 46m 17s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/726e0b53aaa34a1586a4b89835b97f9d : SUCCESS in 39m 36s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6b56dc3deae5429b89d34151a5a64ede : SUCCESS in 43m 04s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/eabe095e88ca44f3b1ca264b1bf0ad2c : SUCCESS in 45m 19s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/fa50597b2dc946de947e5facf23acc73 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9f2f59a651e531c6d8678780c9cfb6491719ed51","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-23 15:15:35.000000000","message":"Patch Set 3: Code-Review+1\n\nWe did a similar change in tempest some time ago: https://review.opendev.org/#/c/663345/6/tempest/api/volume/base.py","accounts_in_message":[],"_revision_number":3},{"id":"5c61ddb0249161bc89b9182efd4adb59a2499a6d","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-09-23 16:01:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aac8985a2f025ec682b766cf8df1de009b53af9f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-23 16:11:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"17803eb8d08786f51fddf97e7027ed0743758086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:11:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"30884f0deb6a6ca1b979ea921121d3eda0002780","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 17:05:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c9b1d78413ee2445eef106ab0f6026a57f6d9019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 17:05:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa89361d02304c579d6eee11824f6dd0 : SUCCESS in 3m 18s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ed074d09e031421ab2ff74a92f2c8005 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da01199b5e6a5fee935c035954d28ab2b3591122","revisions":{"7c1bbd690a865953cc843a74b7ee775bbe3ab067":{"kind":"REWORK","_number":1,"created":"2020-09-12 01:01:12.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/59/751559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/751559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/1"}}},"commit":{"parents":[{"commit":"a10345f9930245cf128261f90c85c73ff913018b","subject":"Merge \"Use the encryption provider constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a10345f9930245cf128261f90c85c73ff913018b"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 00:32:23.000000000","tz":120},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 01:01:09.000000000","tz":120},"subject":"Ceph Multibackend chanages","message":"Ceph Multibackend chanages\n\n1. Tempest has an option CONF.compute.compute_volume_common_az to make\nsure that Volume and Instance are created in common AZ If Availability\nzones are used. The paramater is respected when instance is created\nbecause tempest.common.compute.create_test_server() is used but not\nrespected when volume is created for test purpose. Let\u0027s make sure\nAZ is set when volume is created if parameter is set.\n\n2. The plain Encryptor provider is not support on ceph backends, the\ntest test_encrypted_cinder_volumes_cryptsetup fails on:\ncinder.exception.VolumeDriverException: Volume driver reported an error:\nProvider plain not supported.\nThis comes from [1] which is called in [2]. The Plain provider will be deprecated\nin future completely.\n\n[1] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/volume_utils.py#L1195\n[2] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/drivers/rbd.py#L877\n\nChange-Id: Ib6290ed1ad5779e4e6fed449f3335dbdada42f0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7c1bbd690a865953cc843a74b7ee775bbe3ab067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7c1bbd690a865953cc843a74b7ee775bbe3ab067"}]},"branch":"refs/heads/master"},"7f915fbe4c33f79f51abab18ebde8cfaebb5c247":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-12 01:02:48.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/59/751559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/751559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/2"}}},"commit":{"parents":[{"commit":"a10345f9930245cf128261f90c85c73ff913018b","subject":"Merge \"Use the encryption provider constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a10345f9930245cf128261f90c85c73ff913018b"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 00:32:23.000000000","tz":120},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 01:02:33.000000000","tz":120},"subject":"Ceph Multibackend changes","message":"Ceph Multibackend changes\n\n1. Tempest has an option CONF.compute.compute_volume_common_az to make\nsure that Volume and Instance are created in common AZ If Availability\nzones are used. The paramater is respected when instance is created\nbecause tempest.common.compute.create_test_server() is used but not\nrespected when volume is created for test purpose. Let\u0027s make sure\nAZ is set when volume is created if parameter is set.\n\n2. The plain Encryptor provider is not support on ceph backends, the\ntest test_encrypted_cinder_volumes_cryptsetup fails on:\ncinder.exception.VolumeDriverException: Volume driver reported an error:\nProvider plain not supported.\nThis comes from [1] which is called in [2]. The Plain provider will be deprecated\nin future completely.\n\n[1] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/volume_utils.py#L1195\n[2] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/drivers/rbd.py#L877\n\nChange-Id: Ib6290ed1ad5779e4e6fed449f3335dbdada42f0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7f915fbe4c33f79f51abab18ebde8cfaebb5c247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7f915fbe4c33f79f51abab18ebde8cfaebb5c247"}]},"branch":"refs/heads/master"},"da01199b5e6a5fee935c035954d28ab2b3591122":{"kind":"REWORK","_number":3,"created":"2020-09-12 01:04:30.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/59/751559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/59/751559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/59/751559/3"}}},"commit":{"parents":[{"commit":"a10345f9930245cf128261f90c85c73ff913018b","subject":"Merge \"Use the encryption provider constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a10345f9930245cf128261f90c85c73ff913018b"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 00:32:23.000000000","tz":120},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-09-12 01:04:04.000000000","tz":120},"subject":"Ceph Multibackend changes","message":"Ceph Multibackend changes\n\n1. Tempest has an option CONF.compute.compute_volume_common_az to make\nsure that Volume and Instance are created in common AZ If Availability\nzones are used. The paramater is respected when instance is created\nbecause tempest.common.compute.create_test_server() is used but not\nrespected when volume is created for test purpose. Let\u0027s make sure\nAZ is set when volume is created if parameter is set.\n\n2. The plain Encryptor provider is not support on ceph backends, the\ntest test_encrypted_cinder_volumes_cryptsetup fails on:\ncinder.exception.VolumeDriverException: Volume driver reported an error:\nProvider plain not supported.\nThis comes from [1] which is called in [2]. The Plain provider will be deprecated\nin future completely.\n\n[1] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/volume_utils.py#L1195\n[2] https://opendev.org/openstack/cinder/src/branch/master/cinder/volume/drivers/rbd.py#L877\n\nChange-Id: Ib6290ed1ad5779e4e6fed449f3335dbdada42f0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da01199b5e6a5fee935c035954d28ab2b3591122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da01199b5e6a5fee935c035954d28ab2b3591122"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
