)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~763485","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I1f0d8be1b237da0c96e820c4b3dca09a83b29752","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"glance-multistore","hashtags":[],"change_id":"I1f0d8be1b237da0c96e820c4b3dca09a83b29752","subject":"Copy created image into all available glance stores","status":"MERGED","created":"2020-11-20 00:33:50.000000000","updated":"2021-02-09 14:55:16.000000000","submitted":"2021-02-09 14:55:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763485-glance-multistore","meta_rev_id":"c45f41a032aa75173678aff5242e5210fea51431","_number":763485,"virtual_id_number":763485,"owner":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 14:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-02-09 14:55:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-02-09 14:55:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-02-09 13:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 04:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 18:52:11.000000000","updated_by":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-12-01 18:52:11.000000000","updated_by":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"af35a28251366c8663f6c81b97189db3f3379c3f","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-11-20 00:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b9382a4befb7c5361f7146b8f58a4b1e4c44c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:38:13.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cd8a9e751945398ad8fe2f0c612182 : FAILURE in 2m 47s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/69a2c42fe5a5418288f0c3e3c5504179 : SUCCESS in 44m 33s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/b77b9fba3b09460c85a8c942117da2dd : SUCCESS in 38m 55s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/9e8298b41f5745e8871fa0e5b85af067 : SUCCESS in 44m 55s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/b26b08948a984ceb93dfec124e73a0c8 : SUCCESS in 39m 32s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/0740d227dc9643cabd01baaf1ef3b8d7 : FAILURE in 47m 44s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ea18bb5e086c4e9199b8d6a623cee165 : SUCCESS in 47m 06s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/cc5b1d2996084be58a2ecdeded49c18a : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c5c2ad1ad30081e476337e775d3079d194d094b6","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-11-20 01:44:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2ea2faf1fe7972b44dbf4c6ec5688f68eeeed6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 02:46:28.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb1a84a7ba5451c9ff9344b879aad47 : FAILURE in 3m 15s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9e131ae534c7451aa01d9a9ee9ef40a3 : SUCCESS in 40m 29s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/7433a287b92d4198af1adea0c06baacb : SUCCESS in 35m 45s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/e540a13ae64c48e2a39cbc22240ef58a : SUCCESS in 37m 18s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/da44eec3f272493aa7528e215b0dd226 : SUCCESS in 36m 19s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/02c115ff94b142d1b71ed6b6236b96c6 : FAILURE in 49m 30s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/658ec82e6a2c4ff4a1c7e67c503346a5 : SUCCESS in 46m 29s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/966ed33bdfd44a7a8a545bf498246aba : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f9be88de68a2231791177e3ebdbd27851a4b9bf7","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-11-20 03:29:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efc9b8a9e308748de912b64dd108b1f173a71f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 04:23:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec2d0a902064149bbdbdec781fcca8a : SUCCESS in 3m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6fe514bdb1b2454db1b939765c6b56b5 : SUCCESS in 40m 39s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/4c558671407846afa4854aa29ba8a6db : SUCCESS in 44m 49s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/60bc9286854a4a82baa817ec2fb8ca87 : SUCCESS in 33m 16s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/09c54726bbe645c2bc0e65dad2ac5dac : SUCCESS in 46m 31s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/fe68efd0f2af4ee3b9c4f9fc6ccddc1d : SUCCESS in 46m 00s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/a11da07079eb4f7392d5643cb5469be8 : FAILURE in 43m 04s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/91cf08f88f6b4994936b5688e2c50e3c : FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f928376c902f645b77210a53772d333306608481","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-11-21 23:12:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf4c6a4151714556d005cedae6e45980bb8c7e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-22 00:06:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee40af4896be4f139381c7d0f3c61a70 : SUCCESS in 2m 26s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3c3fd5db0373414594a212a045581cab : SUCCESS in 43m 39s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/84feda745f024daaa13ea9759f748343 : SUCCESS in 42m 10s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/b850ef9c84f944baa0eb8ad5dad8631c : SUCCESS in 40m 53s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/d72005e9d436449088edba248fc20d1f : SUCCESS in 43m 07s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c23b553e7e7e403982d3d6e6390118e3 : SUCCESS in 43m 48s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/c1398258065745b19c7ada15e7b22a84 : SUCCESS in 46m 00s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/fcdd22bfdf0d4ccfa4e006e15557221c : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":3},{"id":"566cd78e9ce817b71d312dc08dceb68675ec3e0e","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-12-02 18:42:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b0895b563226aef08ccfb9b2e419a4dbf82322b0","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-12-08 13:20:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff8181fe42af0fe8e10c2c631af21d5bd3704985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 13:21:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"82629ce4b03baf237757dbcfbb99291632eb9d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 14:17:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f39ede14b0e40819ec793e8dd570ee5 : FAILURE in 2m 33s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5191611ad5614971b21c11fdfca500ee : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b826ba0db96ebec1591c07aedbb6257d9267c949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"date":"2020-12-09 07:48:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d533225e8df21dbda0ded873dc9e0ef2c14fd704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 08:42:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18a5d8b795d47d28708552b983ecae7 : FAILURE in 2m 27s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/72d86a79ad734b7b9e77196b9dab3584 : SUCCESS in 47m 56s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/44e9ca825f7d4ad0b3c5792cc0309d25 : SUCCESS in 44m 55s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/f142abecbe884bf49d64b785020c5e6f : FAILURE in 18m 37s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/6415ef85aa49405c8e6c5361068ec05b : FAILURE in 16m 38s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9a0ad1d1f56440ba8c7b761e4cd5bc5a : FAILURE in 41m 24s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/39ca387844b144e398635eef6b73c4dd : SUCCESS in 44m 22s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/556e443e8a5242a98ac7a23e8c0f86dc : SUCCESS in 42m 29s","accounts_in_message":[],"_revision_number":4},{"id":"59ddbe4e0064c2c426f6a5a4b320f91cbdb7a392","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-01-21 16:13:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"31b335e78bedafbb83262a8bbbc58b7ef70a8d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:22:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7da29587b044b485a50b7099b86c87 : FAILURE in 4m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2b2545eff8284257aeb8fe09777cae78 : FAILURE in 48m 04s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e75bd1f9af044fe4b09531dc08658c6b : SUCCESS in 55m 18s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/bbe75df4593f45ac940ca52cf35a6200 : SUCCESS in 39m 42s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/b0b9323d74144538b051d3cb40a1032b : SUCCESS in 37m 27s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d3ea4f338cf745539f4546ba8ee3971a : FAILURE in 50m 42s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/a8c4e2f1fd204a679c23daa7119ed847 : SUCCESS in 39m 17s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/89cd5c876126471d9a8fe644e853c48e : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":4},{"id":"df4ef489dbaf009adfae0c21782012bba5f29160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-02-08 23:00:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b8d64e412e2051a3d3a03418c969ac38df60be46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:53:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce898f5e9705493e8286da972d41e506 : SUCCESS in 3m 10s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e5ee141c950643f1b0bfb9d5baa3b82f : SUCCESS in 39m 09s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/d7f833a6b7054f53809b69d0abca5d38 : SUCCESS in 39m 02s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/2fc9bfe811ed494cad4e64a88b8ad832 : SUCCESS in 42m 54s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/8510c7ef2ed94ad7ae1d8fe50e0137a2 : SUCCESS in 41m 27s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/98b7053ee99541cebd5d325c7cf747fd : SUCCESS in 38m 30s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/e9ffd1699b8c4410a437ea3f5c77968e : SUCCESS in 39m 14s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/7696c86c4c834138b42dd441fd48f838 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":5},{"id":"09d6c8477beb824f0a775070bf8b6cbb2c985fae","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-02-09 13:25:35.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dbd2bee712bba065385239295df5e3cbc0abb429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:25:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"25d658a7fb4ee3bb5a2931bb26b84d27ff44b3e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:55:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c76dfeed2743babd0338d2a5c1f44b : SUCCESS in 3m 38s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cec6efc3c17c4cb7ba92e57d2944d01f : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":5},{"id":"0974583b021e29e1a7b2921f0f80dfff3916a9bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:55:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1972c4692428a61fafc7e1290632395d5824a4d8","revisions":{"b9e0a6e63e4d1a653f7439630db41b851257e695":{"kind":"REWORK","_number":1,"created":"2020-11-20 00:33:50.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/85/763485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/85/763485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/1"}}},"commit":{"parents":[{"commit":"da01199b5e6a5fee935c035954d28ab2b3591122","subject":"Ceph Multibackend changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da01199b5e6a5fee935c035954d28ab2b3591122"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"subject":"Copy created image into all available glance stores","message":"Copy created image into all available glance stores\n\nThe openstack can have multiple glance stores deployed/available.\nIt may be a proper thing to copy newly created signed image into\nall the available glance stores so barbican tempest tests can access\nimage regardless on which compute and storage backend they are ran on\nand regardless on which glance store is local.\nAdditionally there is a nova-compute conf parameter[1] which can even\nprevent instances being spawned from image which is not available in local\nglance store.\nThe copy-image would happen only if import_image tempest cong option\nis available which indicates glance multistore is available.\n\n[1] https://review.opendev.org/#/c/657078/\n\nChange-Id: I1f0d8be1b237da0c96e820c4b3dca09a83b29752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b9e0a6e63e4d1a653f7439630db41b851257e695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b9e0a6e63e4d1a653f7439630db41b851257e695"}]},"branch":"refs/heads/master"},"2ebe3b29e77076e369db6f693a61aa6f1c2d90d4":{"kind":"REWORK","_number":2,"created":"2020-11-20 01:44:11.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/85/763485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/85/763485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/2"}}},"commit":{"parents":[{"commit":"da01199b5e6a5fee935c035954d28ab2b3591122","subject":"Ceph Multibackend changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da01199b5e6a5fee935c035954d28ab2b3591122"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 01:43:39.000000000","tz":60},"subject":"Copy created image into all available glance stores","message":"Copy created image into all available glance stores\n\nThe openstack can have multiple glance stores deployed/available.\nIt may be a proper thing to copy newly created signed image into\nall the available glance stores so barbican tempest tests can access\nimage regardless on which compute and storage backend they are ran on\nand regardless on which glance store is local.\nAdditionally there is a nova-compute conf parameter[1] which can even\nprevent instances being spawned from image which is not available in local\nglance store.\nThe copy-image would happen only if import_image tempest cong option\nis available which indicates glance multistore is available.\n\n[1] https://review.opendev.org/#/c/657078/\n\nChange-Id: I1f0d8be1b237da0c96e820c4b3dca09a83b29752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2ebe3b29e77076e369db6f693a61aa6f1c2d90d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2ebe3b29e77076e369db6f693a61aa6f1c2d90d4"}]},"branch":"refs/heads/master"},"ede87c5c3ecedf772ed50e24009e84f448813d02":{"kind":"REWORK","_number":3,"created":"2020-11-20 03:29:43.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/85/763485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/85/763485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/3"}}},"commit":{"parents":[{"commit":"da01199b5e6a5fee935c035954d28ab2b3591122","subject":"Ceph Multibackend changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/da01199b5e6a5fee935c035954d28ab2b3591122"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 03:29:20.000000000","tz":60},"subject":"Copy created image into all available glance stores","message":"Copy created image into all available glance stores\n\nThe openstack can have multiple glance stores deployed/available.\nIt may be a proper thing to copy newly created signed image into\nall the available glance stores so barbican tempest tests can access\nimage regardless on which compute and storage backend they are ran on\nand regardless on which glance store is local.\nAdditionally there is a nova-compute conf parameter[1] which can even\nprevent instances being spawned from image which is not available in local\nglance store.\nThe copy-image would happen only if import_image tempest cong option\nis available which indicates glance multistore is available.\n\n[1] https://review.opendev.org/#/c/657078/\n\nChange-Id: I1f0d8be1b237da0c96e820c4b3dca09a83b29752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ede87c5c3ecedf772ed50e24009e84f448813d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ede87c5c3ecedf772ed50e24009e84f448813d02"}]},"branch":"refs/heads/master"},"ef84a26d0ff4621f3b1b9fc766813e55eb070ce8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-12-09 07:48:02.000000000","uploader":{"_account_id":15206,"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","username":"mvk"},"ref":"refs/changes/85/763485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/85/763485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/4"}}},"commit":{"parents":[{"commit":"53dcab70f26f8e990fce425ee1005ef3f775d174","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/53dcab70f26f8e990fce425ee1005ef3f775d174"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"committer":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-12-09 07:48:02.000000000","tz":0},"subject":"Copy created image into all available glance stores","message":"Copy created image into all available glance stores\n\nThe openstack can have multiple glance stores deployed/available.\nIt may be a proper thing to copy newly created signed image into\nall the available glance stores so barbican tempest tests can access\nimage regardless on which compute and storage backend they are ran on\nand regardless on which glance store is local.\nAdditionally there is a nova-compute conf parameter[1] which can even\nprevent instances being spawned from image which is not available in local\nglance store.\nThe copy-image would happen only if import_image tempest cong option\nis available which indicates glance multistore is available.\n\n[1] https://review.opendev.org/#/c/657078/\n\nChange-Id: I1f0d8be1b237da0c96e820c4b3dca09a83b29752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ef84a26d0ff4621f3b1b9fc766813e55eb070ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ef84a26d0ff4621f3b1b9fc766813e55eb070ce8"}]},"branch":"refs/heads/master","description":"Rebase"},"1972c4692428a61fafc7e1290632395d5824a4d8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-08 23:00:39.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/85/763485/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/85/763485/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/85/763485/5"}}},"commit":{"parents":[{"commit":"a318f6d77c0b354b4aee2fe9d7651d595adcf38e","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a318f6d77c0b354b4aee2fe9d7651d595adcf38e"}]}],"author":{"name":"Marian Krcmarik","email":"mkrcmari@redhat.com","date":"2020-11-20 00:33:02.000000000","tz":60},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-02-08 23:00:39.000000000","tz":0},"subject":"Copy created image into all available glance stores","message":"Copy created image into all available glance stores\n\nThe openstack can have multiple glance stores deployed/available.\nIt may be a proper thing to copy newly created signed image into\nall the available glance stores so barbican tempest tests can access\nimage regardless on which compute and storage backend they are ran on\nand regardless on which glance store is local.\nAdditionally there is a nova-compute conf parameter[1] which can even\nprevent instances being spawned from image which is not available in local\nglance store.\nThe copy-image would happen only if import_image tempest cong option\nis available which indicates glance multistore is available.\n\n[1] https://review.opendev.org/#/c/657078/\n\nChange-Id: I1f0d8be1b237da0c96e820c4b3dca09a83b29752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1972c4692428a61fafc7e1290632395d5824a4d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1972c4692428a61fafc7e1290632395d5824a4d8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
