)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~774771","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ib79eed6886839d1b7848c991bd64e82595c6c32e","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"add_rbac_tests","hashtags":[],"change_id":"Ib79eed6886839d1b7848c991bd64e82595c6c32e","subject":"Initial patch to add barbican rbac tests","status":"MERGED","created":"2021-02-10 06:15:30.000000000","updated":"2021-03-13 01:57:03.000000000","submitted":"2021-03-13 01:57:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774771-add_rbac_tests","meta_rev_id":"21e100fa4d29dc00025e56d735b8d5a1148838ff","_number":774771,"virtual_id_number":774771,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 01:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-03-13 01:22:03.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-03-13 01:57:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-03-13 01:22:03.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}],"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 07:06:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-15 21:07:52.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-03-11 20:07:38.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-11 20:07:38.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"5275f616feb04310b0a720068d018c3832ba1ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-02-10 06:15:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d0e14994e90ca708ae30ccb925c867217e7ca88","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-02-10 06:15:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"261a9e8599e9461adbab0c44026ec4fbc0506762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 07:06:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b83699eb4c467ebc057636425b40b1 : SUCCESS in 2m 51s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8e5978fbcda44bc8928bddbdb777cd95 : FAILURE in 35m 43s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/31785099f2624328bc2010b82dddb5d0 : FAILURE in 42m 04s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/3e09863a4e704f89998a1cdf0d807956 : FAILURE in 39m 28s\n- barbican-tempest-plugin-simple-crypto-stein https://zuul.opendev.org/t/openstack/build/db20c659cf6446aea0e9094de478e46b : FAILURE in 30m 21s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/14924f95bf99487e83235d3d29594819 : FAILURE in 33m 37s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6149e4334240497bae978396ae53a416 : FAILURE in 34m 34s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/3a3ef719c0564a118c69e8580735b571 : FAILURE in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"423bd289f860fce75f36e7be2f8aa5794ed3572f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-02-15 20:51:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08cc0f64f8dafd361782b7598b081c7c535d6b02","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-15 21:07:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08d516a6596982a66c458e61726df58a36172d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 22:13:05.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e8ff912336452b8ce1555a2789c8d6 : FAILURE in 3m 13s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a84fc35f64164c558f95bbdc1b1840d0 : SUCCESS in 33m 00s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/76c3c3c2641e4350a7661311d8194624 : SUCCESS in 37m 53s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/336ba84dc4e14c2aa1080d857713ff89 : SUCCESS in 45m 26s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/607c3f5429054c099bedd835c4690d5b : SUCCESS in 39m 12s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/1e82e69954214f30bc6324913f41ec5b : SUCCESS in 54m 13s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/366e05cfe77344df8777c57dc1f48669 : SUCCESS in 44m 42s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/1c167a9b47404a829346a75868262842 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":2},{"id":"cf5c01f7e0b4cb39f44af8746a5a676c3b20e928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-02-15 23:23:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"edc340c11ebc2d157659ce052a6a1ee24d228614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 00:20:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ffee2cfac784293a5f1d21d09b0a1ec : SUCCESS in 4m 22s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fc61ce648c034f8ab7940473c9207706 : SUCCESS in 38m 28s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/18b3fe00a3c04a3bb01c3603541af264 : SUCCESS in 38m 04s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f4cef1fe14054889bdfba15c81cd1757 : SUCCESS in 37m 21s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/5a97f9b4d3144990b30f06fac76f935e : SUCCESS in 37m 59s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/de2e3b214d8544e9b4a1e24e6a481519 : SUCCESS in 39m 21s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/14fd0028dc834b09832bd640a7cc608b : SUCCESS in 39m 34s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f5559ae5ba18497c846fb873fa7bcd55 : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":3},{"id":"46d2d30032e9ce89c21bb65ca73517d51fb8bb26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-11 21:42:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"dbe301c1baf95f112b786d0fd847869fcb161b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:33:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d2746eed3b4edebfcbc56eb36d786a : SUCCESS in 3m 33s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/23658b15427745ddb80d048f29f04ab1 : SUCCESS in 40m 12s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/68f336e61b9c453e96a12a8538deb79d : SUCCESS in 39m 57s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/7ca1af01e62348db88f3b27378891084 : SUCCESS in 41m 05s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/5b8961334bae442ca8960aa77ae3086a : SUCCESS in 40m 48s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/dc2ee1d8ad27402dbd0b17017c0c7016 : SUCCESS in 46m 06s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/9cfa5f0f59524f9c811356226cb13103 : SUCCESS in 40m 38s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/02a2543cacc840dca0a95255c69d52fa : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":4},{"id":"4fd07800c9311ce3c42a461000739bce0908b5f7","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-12 16:55:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"29fe95e034b194fb9dfce9175ffa0d19efe1938d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-12 18:05:26.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"09be846cec2a094067fa31f098096a658147203b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:05:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fad619f3f3247fc8b10a36155c40164 : SUCCESS in 3m 52s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7fafaf966f15438589ce47c4bc22885d : SUCCESS in 44m 03s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/8e9e0a5de23f4598bf3c6fae22d91586 : SUCCESS in 42m 46s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/3009c5356c144ae281b49ff006ba6a02 : SUCCESS in 39m 48s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/f335b714da2448b2b554771591dfd912 : SUCCESS in 38m 17s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/7c511b83b2e648669f016d6159535235 : SUCCESS in 51m 39s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/06b9d8b7629f446486db8f51e04e3afa : SUCCESS in 46m 50s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ca248311dee447ee985139ae0240a8bc : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":5},{"id":"e7fe48e4efc601a7f54b947bd4572356f8de6c5a","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-13 01:22:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"34409f8ecab343968b3f5c9ca955ea89c12cd0be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:22:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fe630bd8516e732b8a2b9980477f377fd4a581f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:57:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50fd663d10e40a0bbae2f84975c8476 : SUCCESS in 2m 40s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e5b72591a9ca45f394df6309ccecb868 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":5},{"id":"f7b625d95964d9713e4e9196d26637eaecdc311a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:57:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c80b00af9334df2f57d9e56804a9c807a44749bc","revisions":{"652fbbaa4181d9dd29bde4d3f63c75e9d226ff99":{"kind":"REWORK","_number":1,"created":"2021-02-10 06:15:30.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/774771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/774771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/1"}}},"commit":{"parents":[{"commit":"53dcab70f26f8e990fce425ee1005ef3f775d174","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/53dcab70f26f8e990fce425ee1005ef3f775d174"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-09 19:49:41.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-10 06:15:27.000000000","tz":-300},"subject":"Initial patch to add barbican rbac tests","message":"Initial patch to add barbican rbac tests\n\nThis adds initial RBAC tests for secrets\n\nChange-Id: Ib79eed6886839d1b7848c991bd64e82595c6c32e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/652fbbaa4181d9dd29bde4d3f63c75e9d226ff99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/652fbbaa4181d9dd29bde4d3f63c75e9d226ff99"}]},"branch":"refs/heads/master"},"1fab6d8ce7942e12d8fc5446fdc94846e197415b":{"kind":"REWORK","_number":2,"created":"2021-02-15 20:51:06.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/774771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/774771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/2"}}},"commit":{"parents":[{"commit":"53dcab70f26f8e990fce425ee1005ef3f775d174","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/53dcab70f26f8e990fce425ee1005ef3f775d174"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-09 19:49:41.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-15 20:50:56.000000000","tz":-300},"subject":"Initial patch to add barbican rbac tests","message":"Initial patch to add barbican rbac tests\n\nThis adds initial RBAC tests for secrets\n\nChange-Id: Ib79eed6886839d1b7848c991bd64e82595c6c32e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1fab6d8ce7942e12d8fc5446fdc94846e197415b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1fab6d8ce7942e12d8fc5446fdc94846e197415b"}]},"branch":"refs/heads/master"},"5fc2375ffb0175b56a2004d2cd7d15d6be108c8d":{"kind":"REWORK","_number":3,"created":"2021-02-15 23:23:41.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/774771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/774771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/3"}}},"commit":{"parents":[{"commit":"53dcab70f26f8e990fce425ee1005ef3f775d174","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/53dcab70f26f8e990fce425ee1005ef3f775d174"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-09 19:49:41.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-15 23:23:34.000000000","tz":-300},"subject":"Initial patch to add barbican rbac tests","message":"Initial patch to add barbican rbac tests\n\nThis adds initial RBAC tests for secrets\n\nChange-Id: Ib79eed6886839d1b7848c991bd64e82595c6c32e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5fc2375ffb0175b56a2004d2cd7d15d6be108c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5fc2375ffb0175b56a2004d2cd7d15d6be108c8d"}]},"branch":"refs/heads/master"},"116a6fadb2ef28c23e295f1fa9b7eb2976deb9e5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-11 21:42:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/774771/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/774771/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/4"}}},"commit":{"parents":[{"commit":"ad7f742ea9edd5508123b321d46ec2f4e1af8271","subject":"Merge \"Fix test_signed_image_upload_boot_failure skip method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ad7f742ea9edd5508123b321d46ec2f4e1af8271"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-09 19:49:41.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-11 21:42:07.000000000","tz":0},"subject":"Initial patch to add barbican rbac tests","message":"Initial patch to add barbican rbac tests\n\nThis adds initial RBAC tests for secrets\n\nChange-Id: Ib79eed6886839d1b7848c991bd64e82595c6c32e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/116a6fadb2ef28c23e295f1fa9b7eb2976deb9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/116a6fadb2ef28c23e295f1fa9b7eb2976deb9e5"}]},"branch":"refs/heads/master","description":"Rebase"},"c80b00af9334df2f57d9e56804a9c807a44749bc":{"kind":"NO_CHANGE","_number":5,"created":"2021-03-12 18:05:26.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/774771/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/774771/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/774771/5"}}},"commit":{"parents":[{"commit":"ad7f742ea9edd5508123b321d46ec2f4e1af8271","subject":"Merge \"Fix test_signed_image_upload_boot_failure skip method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ad7f742ea9edd5508123b321d46ec2f4e1af8271"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-09 19:49:41.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-12 18:03:24.000000000","tz":-300},"subject":"Initial patch to add barbican rbac tests","message":"Initial patch to add barbican rbac tests\n\nThis adds initial RBAC tests for secrets\n\nChange-Id: Ib79eed6886839d1b7848c991bd64e82595c6c32e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c80b00af9334df2f57d9e56804a9c807a44749bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c80b00af9334df2f57d9e56804a9c807a44749bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
