)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~784160","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I2272f5cd2385df15cc7761e8b858dfd39be955d4","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I2272f5cd2385df15cc7761e8b858dfd39be955d4","subject":"Add secure-rbac tests for Containers","status":"MERGED","created":"2021-03-31 19:57:40.000000000","updated":"2021-04-09 21:20:41.000000000","submitted":"2021-04-07 09:25:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784160-secure-rbac","meta_rev_id":"42d91ff49c4e645b776f97867e15b7bf0feedefa","_number":784160,"virtual_id_number":784160,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 09:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-04-07 08:28:58.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-04-05 15:53:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-04-07 08:28:58.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 19:58:08.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-03-31 19:58:08.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-31 19:58:08.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-03-31 21:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 15:40:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"8756dca73698b77f01674e16110e7462f0f717e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-31 19:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6202276d7b98cca1f9c0eb194d772a004b912264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-31 20:05:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f39405152278540421bd63572478727b9b0508c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 21:44:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1f1f000e33471d88e56b9b435cd044 : SUCCESS in 16m 21s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c74322543e354a64b27195529bd90f48 : SUCCESS in 50m 08s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/48153c12a5504596b9756ce861ac567a : FAILURE in 40m 40s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/a87c11054e6a4c7cacba4c11318183c8 : SUCCESS in 39m 54s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/8c3e9f314dee4ca3862db83cc1969991 : SUCCESS in 40m 11s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/04233be50fbb455f8e31a7d3261f806f : SUCCESS in 43m 45s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c516b94766324acaa22f52a402ca0dc6 : SUCCESS in 50m 54s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/d93e326e369143d38153d96f2b53500b : SUCCESS in 43m 59s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/56bd4774ee4b47ff86ba8b02f8b51b6c : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":2},{"id":"537b9e52e4312381151edca479746eda796ca332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-01 13:36:32.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e922563dd72605f7d51fc58e1cec734ca1ce2f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:37:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf74d46f52e44999bb89cac43f585c8b : SUCCESS in 3m 57s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8a792426bbd74263ad005c36f9f13683 : SUCCESS in 45m 03s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f8c44b3c9e1645e2bb2776558f2a3c9a : SUCCESS in 49m 01s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/90cf3693ba9a472d8c1ae1d82ee561a0 : SUCCESS in 47m 09s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/a0580ce00f8f49d083175bc86a7f6775 : SUCCESS in 49m 01s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/ee23064cbc9e42b6b2f29a3762a7e446 : SUCCESS in 27m 10s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/2a3fd45d0f9e46b7820aec1895f2a642 : SUCCESS in 52m 27s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/3a548493da8f4b2086e063c8cddbd0c3 : SUCCESS in 55m 12s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/9f0564848bef41ceb85c67773577208b : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0574451029150d4ab717ee823166cf90d2777c9b","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-04-01 20:36:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"2114f2af254245d1243e023f8904bb051d97c322","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-01 21:54:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd31770a0cb7b4ddc53f121e7ffc5734bfaf42bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-01 22:05:55.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ab2a93833274b92deec20a41b25c3c18162537c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 23:09:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4815e212f5f4174ab30c77db01b2626 : SUCCESS in 3m 43s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1237fd48e864446d9e0941c57179a2b1 : SUCCESS in 42m 44s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/94e1db577f304ecd9bc65956d93d9b0b : SUCCESS in 44m 06s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/ab5fd35927d4485e8bf58c2b2a44ef84 : SUCCESS in 50m 50s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/12999a42545f48d18e39869c4fbfa526 : SUCCESS in 46m 55s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/59da629ca63b4d168aedcee14597752a : SUCCESS in 49m 37s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9b654025b3fa4421bec88f93cce6c92e : SUCCESS in 55m 12s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/af02a294762a44498fddee5240b4e04e : SUCCESS in 46m 58s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/5615d68e123c4e30a5ae0504d5eb89e8 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0baf8011f3bff0a03caeae42ea9c25e9fea7728c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-04-05 15:53:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"938ccfe0f9df39eed224837e697f742075cb204a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-05 18:14:55.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nLooks good - just a few minor suggestions inline to increase protection coverage for multi-tenancy.","accounts_in_message":[],"_revision_number":4},{"id":"2b5a74ce3761c0b48da9f49a99c481dcdf0befd5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-06 13:35:18.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"171ced235cfe538a75bbbb5126bcc7cd7a5d162a","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-04-07 08:28:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"858d8f0fde2721b47a477cb7036a136f2632f087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 08:29:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6c490e476616f2edbb5f652d2889329f3d5fce44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:25:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a826c2a302f401ca9654c7c7e7d568e : SUCCESS in 4m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/07fbd9b9fbcd46c2b95f528e5223aa73 : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":4},{"id":"4cbc85fd659959d10ecb06c12e2325f7e8c5d2a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:25:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"42d91ff49c4e645b776f97867e15b7bf0feedefa","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-09 21:20:41.000000000","message":"Patch Set 4:\n\nAck - thanks for following up my comments, Doug.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"db9b512ead907f42a8a767a73cc44a240437e542","revisions":{"62d706e77549375ce152bd134353ebbb83664d9b":{"kind":"REWORK","_number":1,"created":"2021-03-31 19:57:40.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/60/784160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/60/784160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/1"}}},"commit":{"parents":[{"commit":"d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b","subject":"Add secure-rbac gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 19:55:54.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 19:57:38.000000000","tz":-300},"subject":"Add secure-rbac tests for Containers","message":"Add secure-rbac tests for Containers\n\nThis patch adds basic RBAC testing for the Containers resource for\nreader, member and admin personas.\n\nDepends-On: Ib780fe16282ea338e98c9c34b31d80bebafdcf52\nChange-Id: I2272f5cd2385df15cc7761e8b858dfd39be955d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62d706e77549375ce152bd134353ebbb83664d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62d706e77549375ce152bd134353ebbb83664d9b"}]},"branch":"refs/heads/master"},"26748be380e39618c22b180c2b06167294912783":{"kind":"REWORK","_number":2,"created":"2021-03-31 20:05:53.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/60/784160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/60/784160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/2"}}},"commit":{"parents":[{"commit":"d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b","subject":"Add secure-rbac gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 19:55:54.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 20:05:46.000000000","tz":-300},"subject":"Add secure-rbac tests for Containers","message":"Add secure-rbac tests for Containers\n\nThis patch adds basic RBAC testing for the Containers resource for\nreader, member and admin personas.\n\nDepends-On: Ib780fe16282ea338e98c9c34b31d80bebafdcf52\nChange-Id: I2272f5cd2385df15cc7761e8b858dfd39be955d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/26748be380e39618c22b180c2b06167294912783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/26748be380e39618c22b180c2b06167294912783"}]},"branch":"refs/heads/master"},"01e76036127e36d7b983b39144cfea3f47a01827":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-01 13:36:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/60/784160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/60/784160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/3"}}},"commit":{"parents":[{"commit":"d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b","subject":"Add secure-rbac gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 19:55:54.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-01 13:35:56.000000000","tz":-300},"subject":"Add secure-rbac tests for Containers","message":"Add secure-rbac tests for Containers\n\nThis patch adds basic RBAC testing for the Containers resource for\nreader, member and admin personas.\n\nDepends-On: I6879f566117db5ec0099ddad35ba649a3c674bd1\nChange-Id: I2272f5cd2385df15cc7761e8b858dfd39be955d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01e76036127e36d7b983b39144cfea3f47a01827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01e76036127e36d7b983b39144cfea3f47a01827"}]},"branch":"refs/heads/master"},"db9b512ead907f42a8a767a73cc44a240437e542":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-01 22:05:55.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/60/784160/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/60/784160/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/60/784160/4"}}},"commit":{"parents":[{"commit":"d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b","subject":"Add secure-rbac gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d013596f3dd03ed11ee9b7e0367dbbbd1df60e4b"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-31 19:55:54.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-01 21:54:25.000000000","tz":-300},"subject":"Add secure-rbac tests for Containers","message":"Add secure-rbac tests for Containers\n\nThis patch adds basic RBAC testing for the Containers resource for\nreader, member and admin personas with project scope.\n\nDepends-On: I6879f566117db5ec0099ddad35ba649a3c674bd1\nChange-Id: I2272f5cd2385df15cc7761e8b858dfd39be955d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/db9b512ead907f42a8a767a73cc44a240437e542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/db9b512ead907f42a8a767a73cc44a240437e542"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
