)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~798008","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I863ca223d6732b43dd6baa1510d4206b34bfb19e","project":"openstack/barbican-tempest-plugin","branch":"master","hashtags":[],"change_id":"I863ca223d6732b43dd6baa1510d4206b34bfb19e","subject":"Let the orders complete before using them","status":"MERGED","created":"2021-06-24 18:59:23.000000000","updated":"2021-08-05 18:50:07.000000000","submitted":"2021-08-05 18:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798008","meta_rev_id":"572334890e0c1960f4a3656d8527285de5b00be6","_number":798008,"virtual_id_number":798008,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-05 18:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2021-08-05 18:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2021-08-05 18:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 19:02:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},"state":"REVIEWER"},{"updated":"2021-06-24 19:02:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-06-24 19:02:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-06-24 19:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bf858f393b6c7444791d1cc74d485c5e8a96353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-24 18:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ae56e19ddc7b5b245201113543894c50bafc7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16ec9b96d084270a9197764ed83dfed : SUCCESS in 5m 22s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1ed1cff02226410cb9c31bd50f5492a7 : SUCCESS in 35m 52s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/cbf024200cc74d8f90ac287c7dd0a08f : SUCCESS in 37m 14s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/a9cd9998d217433eb507c9ef23d5466f : SUCCESS in 46m 41s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/76d67b42474d4179bef378480caeb594 : SUCCESS in 39m 49s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f83858803eb04b958ce6be5ca2b43e43 : SUCCESS in 45m 15s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/6e885ca6526d4bf5bde206e9e5ab4c5b : SUCCESS in 43m 46s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d9a9b0a322624d7da6b6d07e44759cbb : SUCCESS in 44m 16s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/980612bf116749e3badbed90e17eccd4 : SUCCESS in 42m 27s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f266142d385e4924aa397196bd03e32f : SUCCESS in 45m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dd128ea401321bc17a26d3ea0a8d8b4ad8adc925","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-25 13:17:38.000000000","message":"Patch Set 1:\n\nI know I\u0027m going to quote Mr.Obvious, but isn\u0027t a blind sleep a big fragile?","accounts_in_message":[],"_revision_number":1},{"id":"38772a7fa61098e6ba3dd27585897ddaca0fa8d1","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-25 13:46:09.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e I know I\u0027m going to quote Mr.Obvious, but isn\u0027t a blind sleep a big fragile?\n\nACK.  Feels like we could do a little better - we can certainly poll the status of the order.","accounts_in_message":[],"_revision_number":1},{"id":"02e8174c30884deb0e72e59fea788959bda46cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-07-15 22:19:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"451f7e079aa923c315a1640b3993b22a08a9e926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 23:11:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9044d81805b644688a3f6f2191cdb9f5 : SUCCESS in 3m 14s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8fdf080fd70544e19c6ea456a2256346 : SUCCESS in 39m 17s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/06505399891a4111bdb5b8d54854d80a : SUCCESS in 36m 12s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/e801162a7bb948f4867fb8cd9cc13502 : SUCCESS in 46m 10s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/800d6e7af1b94e4dae22e363d68af72d : SUCCESS in 45m 16s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/83976561954b402e9f89158655f31b40 : SUCCESS in 47m 11s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/d43298754fd64b6bb9e7da1d28593b0f : SUCCESS in 43m 33s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ae195c709486424399ca848c6a4153b9 : SUCCESS in 39m 28s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/e3339892882e4d1fa5eb7320df2bed70 : SUCCESS in 39m 13s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/bb720d28f30d4687bd525f935e43f8b8 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"233c0fc5f3dd31b755ab51831d6f95157679dd19","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-05 18:09:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"7d4dcaf3f29844dcc91f16b1b692b34cb8012e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:10:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"efd1301ea179f91d6b60060f0cae5376820a8436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:50:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d915a4b75ff4b0daf746e800cd6b49f : SUCCESS in 3m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/00991a116659404d8fb010f27e00152d : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":2},{"id":"572334890e0c1960f4a3656d8527285de5b00be6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:50:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d4e995e7250715fa8b17daf69ccd5d90de26246","revisions":{"99b4cf4921a0c6d96be25debe82bbf3befd8e23c":{"kind":"REWORK","_number":1,"created":"2021-06-24 18:59:23.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/08/798008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/08/798008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/1"}}},"commit":{"parents":[{"commit":"25e16b8a0f36fbfee8fcccd6926682f0958bcc8d","subject":"Don\u0027t use legacy exponent for RSA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/25e16b8a0f36fbfee8fcccd6926682f0958bcc8d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-06-24 18:58:09.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-06-24 18:58:09.000000000","tz":-300},"subject":"Let the orders complete before using them","message":"Let the orders complete before using them\n\nThis patch changes a test to wait a few seconds for the orders to be\nprocessed by Barbican.  Currently there are intermittent failures when\nthe test executes faster than Barbican can process orders.\n\nChange-Id: I863ca223d6732b43dd6baa1510d4206b34bfb19e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/99b4cf4921a0c6d96be25debe82bbf3befd8e23c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/99b4cf4921a0c6d96be25debe82bbf3befd8e23c"}]},"branch":"refs/heads/master"},"6d4e995e7250715fa8b17daf69ccd5d90de26246":{"kind":"REWORK","_number":2,"created":"2021-07-15 22:19:11.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/08/798008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/08/798008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/08/798008/2"}}},"commit":{"parents":[{"commit":"25e16b8a0f36fbfee8fcccd6926682f0958bcc8d","subject":"Don\u0027t use legacy exponent for RSA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/25e16b8a0f36fbfee8fcccd6926682f0958bcc8d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-06-24 18:58:09.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-07-15 22:19:00.000000000","tz":0},"subject":"Let the orders complete before using them","message":"Let the orders complete before using them\n\nThis patch changes a test to wait a few seconds for the orders to be\nprocessed by Barbican.  Currently there are intermittent failures when\nthe test executes faster than Barbican can process orders.\n\nChange-Id: I863ca223d6732b43dd6baa1510d4206b34bfb19e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6d4e995e7250715fa8b17daf69ccd5d90de26246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6d4e995e7250715fa8b17daf69ccd5d90de26246"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
