)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~801928","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Id9be832a0f255410bd955d94c32001fec500f32f","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secrets-cleanup","hashtags":[],"change_id":"Id9be832a0f255410bd955d94c32001fec500f32f","subject":"Refactor secret cleanup","status":"MERGED","created":"2021-07-22 22:36:56.000000000","updated":"2021-08-18 13:39:21.000000000","submitted":"2021-08-18 13:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801928-secrets-cleanup","meta_rev_id":"8281a73ac44c69b13831871dacc2b2340ab855ae","_number":801928,"virtual_id_number":801928,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 13:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-08-18 08:09:09.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-08-17 19:35:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-08-18 13:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-08-18 13:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-22 23:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 01:50:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},"state":"REVIEWER"},{"updated":"2021-07-23 01:50:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-07-23 01:50:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-07-23 01:50:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-08-17 18:22:35.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"},"state":"REVIEWER"},{"updated":"2021-08-17 19:35:06.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"19910692c016ffffed9bd6a5694c78c8029735f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-07-22 22:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4182cb6c562718bf7b87fdf3a6736e4c357c13eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 23:30:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2c915fb3da343cd97552e8ef2bd76bd : SUCCESS in 5m 49s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dad07958138c4968b48346ef2eae8c5c : SUCCESS in 34m 23s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/98acb67389114564be2a5acc441a375a : SUCCESS in 30m 09s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/8bde4babae7b4dfd93306804def5bf07 : SUCCESS in 38m 59s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/3f9cc2b51b3c4c6f87fe8f4d2fceb022 : SUCCESS in 36m 16s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e7d8f82a8592449fa6cbdd3350db43d8 : SUCCESS in 50m 36s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/294fc140b11b40d6844d52cf90686bc8 : SUCCESS in 35m 28s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/60ec51cf81524576a7587b93e1457e8b : SUCCESS in 42m 32s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ea511d55a8ac4197857c2832d24774cb : SUCCESS in 43m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/26dc56638fd947a59fbbc19e47fade7d : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"931ca6e53c2ff5d00b30e4616ee9ebf06cfc996c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-08-06 08:13:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nHey Doug, please see comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"8321446c949e35498a18967478a3ef5b852f99ac","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-11 15:29:12.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"de7de13720f91123f3158e9939af12fdb26362d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-11 15:48:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5f73f945810353fbfc01777ddff91134f374810","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-11 15:49:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0d66744d44d8f570e88102151a86ab73c2534e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 16:37:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf9802eba504291a89d4d0972724561 : SUCCESS in 3m 35s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dee76196d15e4bc2a6197800e0fc70a7 : SUCCESS in 33m 47s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/a887e03241bc43faae0309bdf0cb0ccf : SUCCESS in 42m 23s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/02affd7cc96044ec9c2395a7beb842b7 : SUCCESS in 40m 48s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/3c6607e937c84a39bb6e06c3b78c86c6 : SUCCESS in 40m 05s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/901fcd7781ab4b67a208a01e64bbdabf : SUCCESS in 39m 25s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/14bf20b87ad24e50a47572e14292ccb2 : SUCCESS in 39m 56s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d4be806e158d44758039e2813d495820 : SUCCESS in 40m 31s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ab6a9fcbad8547899b3292aa75191a25 : SUCCESS in 40m 06s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/af8b56bc74a74eca8ea77ff40cc52039 : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7bbf4383808f68420d2de1789d841084a6f9a1f0","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-08-12 07:50:54.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nfew more comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"5c5c67634caef4e9aa0c06c7099c620d1a1b9a1e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-17 18:03:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af71e99903571e224815b19ee2f3e258e267b206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-17 18:14:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28fb518816912a7f2c9ee5ddb5b1dcd0a743e80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:12:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c9b601c2e04919bb718cf36ef0989f : SUCCESS in 5m 10s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/75271c84c5154a609e7de492745a2f16 : SUCCESS in 40m 13s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/3e6533ef54934da783e34280eed13408 : SUCCESS in 29m 01s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/e2e7ac02ddb4469393c8bf9ee1c58db4 : SUCCESS in 53m 56s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/b1c752f664614a0490b6c3f8c357201d : SUCCESS in 47m 17s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f589efa78d7941b3b06ed3ba464619f2 : SUCCESS in 41m 11s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/85f545e3b5ce4a7ab62c5ac484e485ef : SUCCESS in 45m 38s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/18fc798f51174569859d4c7bfeac8f07 : SUCCESS in 34m 14s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/c73922a2875f4b10b3e1bce26491e041 : SUCCESS in 41m 39s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f0e20b5530e84a8bb5523d66b3df9c9e : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7d8f1c41c945f3c47ab4726fbd42f2acd2cb0f23","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-08-17 19:35:06.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good to me, I think this may actually be a bit more readable.","accounts_in_message":[],"_revision_number":3},{"id":"832d808ddd5e68206f9ac981573d91613ada6b3f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-08-18 08:09:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"db9637cf2762169f7b19e92172e403a0dc126446","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-18 13:00:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2ae57acf96cc721d2d13572ff770417491d791e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 13:00:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f86dfafa477a649e87950ea0eb090b30dfc865d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 13:39:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db95d37646ac491b960c8919295387f1 : SUCCESS in 4m 13s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/94b8019d178f4097922a6b20b4e7e1a5 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8281a73ac44c69b13831871dacc2b2340ab855ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 13:39:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62baff3fccf0a04b4c1c152163477633fa23da81","revisions":{"73297b790e811c184ceb5559c2646b4c577b290c":{"kind":"REWORK","_number":1,"created":"2021-07-22 22:36:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/28/801928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/28/801928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/1"}}},"commit":{"parents":[{"commit":"6d4e995e7250715fa8b17daf69ccd5d90de26246","subject":"Let the orders complete before using them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6d4e995e7250715fa8b17daf69ccd5d90de26246"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-07-22 22:22:00.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-07-22 22:22:00.000000000","tz":0},"subject":"Refactor secret cleanup","message":"Refactor secret cleanup\n\nThis patch is the first in a refactor of the cleanup logic in our\ntests.\n\nThis patch adds a new `cleanup()` method to the SecretClient that\nattempts to delete all the secrets it creates.\n\nMoving the responsibility of tracking which secrets to clean up down\nto the client allows us more flexibility when cleaning up the resources.\ne.g.  it should be fairly easy to clean up secrets across multiple projects\nby just calling the new `cleanup()` method on each client used.\n\nThis patch will also allow us to get rid of the overloaded `do_request()`\nmethod that is currently used as a proxy to the client to be able to track\nentities.\n\nThe change also makes the test code more explicit and easier to read.\n\nChange-Id: Id9be832a0f255410bd955d94c32001fec500f32f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/73297b790e811c184ceb5559c2646b4c577b290c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/73297b790e811c184ceb5559c2646b4c577b290c"}]},"branch":"refs/heads/master"},"82d3b1896f506fedaade03495d8e5e032899f83a":{"kind":"REWORK","_number":2,"created":"2021-08-11 15:48:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/28/801928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/28/801928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/2"}}},"commit":{"parents":[{"commit":"6d4e995e7250715fa8b17daf69ccd5d90de26246","subject":"Let the orders complete before using them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6d4e995e7250715fa8b17daf69ccd5d90de26246"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-07-22 22:22:00.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-08-11 15:47:57.000000000","tz":0},"subject":"Refactor secret cleanup","message":"Refactor secret cleanup\n\nThis patch is the first in a refactor of the cleanup logic in our\ntests.\n\nThis patch adds a new `cleanup()` method to the SecretClient that\nattempts to delete all the secrets it creates.\n\nMoving the responsibility of tracking which secrets to clean up down\nto the client allows us more flexibility when cleaning up the resources.\ne.g.  it should be fairly easy to clean up secrets across multiple projects\nby just calling the new `cleanup()` method on each client used.\n\nThis patch will also allow us to get rid of the overloaded `do_request()`\nmethod that is currently used as a proxy to the client to be able to track\nentities.\n\nThe change also makes the test code more explicit and easier to read.\n\nChange-Id: Id9be832a0f255410bd955d94c32001fec500f32f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82d3b1896f506fedaade03495d8e5e032899f83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82d3b1896f506fedaade03495d8e5e032899f83a"}]},"branch":"refs/heads/master"},"62baff3fccf0a04b4c1c152163477633fa23da81":{"kind":"REWORK","_number":3,"created":"2021-08-17 18:14:45.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/28/801928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/28/801928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/28/801928/3"}}},"commit":{"parents":[{"commit":"6d4e995e7250715fa8b17daf69ccd5d90de26246","subject":"Let the orders complete before using them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6d4e995e7250715fa8b17daf69ccd5d90de26246"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-07-22 22:22:00.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-08-17 18:14:36.000000000","tz":0},"subject":"Refactor secret cleanup","message":"Refactor secret cleanup\n\nThis patch is the first in a refactor of the cleanup logic in our\ntests.\n\nThis patch adds a new `cleanup()` method to the SecretClient that\nattempts to delete all the secrets it creates.\n\nMoving the responsibility of tracking which secrets to clean up down\nto the client allows us more flexibility when cleaning up the resources.\ne.g.  it should be fairly easy to clean up secrets across multiple projects\nby just calling the new `cleanup()` method on each client used.\n\nThis patch will also allow us to get rid of the overloaded `do_request()`\nmethod that is currently used as a proxy to the client to be able to track\nentities.\n\nThe change also makes the test code more explicit and easier to read.\n\nChange-Id: Id9be832a0f255410bd955d94c32001fec500f32f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62baff3fccf0a04b4c1c152163477633fa23da81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/62baff3fccf0a04b4c1c152163477633fa23da81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
