)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~805853","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I80aba2934110965866d1583309df7f2ca9ef4c27","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secure_rbac","hashtags":[],"change_id":"I80aba2934110965866d1583309df7f2ca9ef4c27","subject":"Add secure-rbac test for Containers","status":"MERGED","created":"2021-08-24 13:38:06.000000000","updated":"2021-08-30 17:51:42.000000000","submitted":"2021-08-30 17:51:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"805853-secure_rbac","meta_rev_id":"f2951f0f9048503e8e8f08742bc0b3d137ebbda3","_number":805853,"virtual_id_number":805853,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 17:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-08-30 17:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-08-26 21:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-08-30 17:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 13:38:59.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":33774,"name":"Marek Tamaskovic","email":"mtamasko@redhat.com","username":"mtamasko"},"state":"REVIEWER"},{"updated":"2021-08-24 13:38:59.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-08-24 13:38:59.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-08-24 14:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"571d5666edd5379cf1849fc8e725ce5a6242a71c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-08-24 13:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cb9ed73254f5dd21e35a516b9b06155b7c046f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:34:57.000000000","message":"Patch Set 1: Verified-1\n\n(34 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934344afdfd74ddba14ea2ea828878bc : FAILURE in 3m 49s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f5f2bece124141ce9865c5a9e8dfa45a : SUCCESS in 36m 38s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/1247ea866eb94c62896570558bec1955 : SUCCESS in 32m 05s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/cfe8d8cae50b4891bd1d3713647abd4d : SUCCESS in 53m 09s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/6a78a22061e349dead33f5f7a0cdb2c5 : SUCCESS in 40m 51s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/4e966dbedbca40acb46011805d2ab49c : SUCCESS in 39m 00s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/5f99c3b6ed864e37a30b5e28c71e3594 : SUCCESS in 37m 48s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b66ddaafaaa143f58f93048a3c20943a : SUCCESS in 37m 22s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6595078a0c504b49bd43e4616f967ba9 : SUCCESS in 34m 21s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/843a8b92da074ebbbe340b1cb67c15d2 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"600e9ada7771bce9954265435ec7017b0084703a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-08-24 14:50:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba8bc33674c9f9cc66a9cdcea6d8bf633dd67a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:42:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1ec062031a4d3fb8c5e9943dc82f0c : SUCCESS in 3m 45s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bd6fcbf647a541b39ecb81a0067f7a23 : SUCCESS in 36m 03s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/72bd831ad881420eba613469d1124018 : SUCCESS in 42m 47s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/df35d3825ae94416b9ef4ff829db3070 : SUCCESS in 43m 07s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/2061304f874f4d258a751c8aca9eb4ed : SUCCESS in 40m 30s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/85181e2fddea45aab9db51a3ef39568e : SUCCESS in 45m 40s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/32a236c2cb4648a2a2d150fdb38c05be : SUCCESS in 43m 56s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/14bffcbe82fa4c78acbecf8136061d11 : SUCCESS in 34m 33s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6c522cbe1ca54d0d95be64aff767e11b : SUCCESS in 36m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/741fbc821b2540e19d43d0c4308ebe72 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d1accd472cb355b12253689bdc3d48287ee5de77","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-08-25 15:26:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2378d0ed84723251a8f2cef26d54ad0b92023fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-08-25 16:59:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0b9bc7d65dac198fbfe206a83dcdf51cc5a7300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:51:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458fcb6a1a9d45f0bcaff5d1bdb9c422 : SUCCESS in 5m 46s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/61eadc87a1e5457993d70585386ca0cc : SUCCESS in 40m 09s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/06b4ff289f514935b907b6b672dc2aad : SUCCESS in 37m 22s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/d6f1b83a500c45bdbe287263563a6306 : SUCCESS in 45m 30s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/0e757da7d3e64d72bcaab69ca43c91df : SUCCESS in 38m 43s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/552e8291900e45109ba0b691cd704b70 : SUCCESS in 39m 33s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/30f50468761d43ed90098b6cdcd9c1f1 : SUCCESS in 41m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b94b1d8ddda74313a3f4c2d1321f9044 : SUCCESS in 36m 35s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/2e6cf6c6561c4e65baf4016f82a493a4 : SUCCESS in 39m 44s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/3349c2ad279445eb8c16c8b97aa521ca : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":3},{"id":"dc4a20676e4d25cc3616ca6a64c3c47de2ca0dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-26 18:18:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"dbf72065d3b1a10e63f2a699a80faca34e7418be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 19:09:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4126f173fa91487ab55ce1ff665fc415 : SUCCESS in 3m 38s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/42d7348d19734216a5cf7e12dac80058 : SUCCESS in 43m 07s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/66aec12203744009be5ffce18edf116c : SUCCESS in 34m 12s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/63d2ea298458417789eceb863d0217b7 : SUCCESS in 42m 54s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/b3d9ea2d5b0845359a396c5bfb9980df : SUCCESS in 42m 02s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/0ec87a224cb34b7494bc4bde5ca4ac3d : SUCCESS in 45m 40s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/dc7936b056a64d62905231b174d3a2ac : SUCCESS in 42m 20s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/219c4d523308457599d9243b66513f29 : SUCCESS in 36m 28s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/26dc79bfb38149c1aa81de69f2f4de5b : SUCCESS in 39m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/2c4bd59f0ae4400dbdbf7da3aa59d836 : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":4},{"id":"bfce053ca7b52f76f37905a09dbe95cd242511be","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-26 21:25:30.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c39f1e3bfce1461f0149e93e33d538e33dc826cb","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-30 17:00:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8de692cf6f6cc32485d19ddf1158c50a28d529dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:00:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"244bc13d2295b9b8bb4041d0a72348309233a81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:51:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de144a3f50924cba89fb62221050841b : SUCCESS in 5m 24s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/aeb543166bf84b8cad11233b6a326028 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f2951f0f9048503e8e8f08742bc0b3d137ebbda3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:51:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e5e45748d310f7e944dee17427205b430a954d9","revisions":{"439f6d141924df9f8207ccd0ae17dc43d4b08f95":{"kind":"REWORK","_number":1,"created":"2021-08-24 13:38:06.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/53/805853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/53/805853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/1"}}},"commit":{"parents":[{"commit":"087f89dd16c2fbafaa0aab8e5e2cff09489d40ed","subject":"Add secure-rbac test for Secrets across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/087f89dd16c2fbafaa0aab8e5e2cff09489d40ed"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 13:35:52.000000000","tz":0},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 13:35:52.000000000","tz":0},"subject":"Add secure-rbac test for Containers","message":"Add secure-rbac test for Containers\n\nThis patch adds rbac tests to the Containers resource to test\nsecure-rbac policies within a project\n\nThis patch also removes the use of do_request in the existing tests\nas that method is being deprecated in favor of using the clients\ndirectly.\n\nChange-Id: I80aba2934110965866d1583309df7f2ca9ef4c27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/439f6d141924df9f8207ccd0ae17dc43d4b08f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/439f6d141924df9f8207ccd0ae17dc43d4b08f95"}]},"branch":"refs/heads/master"},"e12adf7c72022ff51352727094e2dcff2a9cb624":{"kind":"REWORK","_number":2,"created":"2021-08-24 14:50:39.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/53/805853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/53/805853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/2"}}},"commit":{"parents":[{"commit":"087f89dd16c2fbafaa0aab8e5e2cff09489d40ed","subject":"Add secure-rbac test for Secrets across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/087f89dd16c2fbafaa0aab8e5e2cff09489d40ed"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 13:35:52.000000000","tz":0},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 14:50:18.000000000","tz":0},"subject":"Add secure-rbac test for Containers","message":"Add secure-rbac test for Containers\n\nThis patch adds rbac tests to the Containers resource to test\nsecure-rbac policies within a project\n\nThis patch also removes the use of do_request in the existing tests\nas that method is being deprecated in favor of using the clients\ndirectly.\n\n* PEP8 Fixes\n\nChange-Id: I80aba2934110965866d1583309df7f2ca9ef4c27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/e12adf7c72022ff51352727094e2dcff2a9cb624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/e12adf7c72022ff51352727094e2dcff2a9cb624"}]},"branch":"refs/heads/master"},"34bcf65ad3ddbbc608382f42cf9fb5e67fa0b439":{"kind":"REWORK","_number":3,"created":"2021-08-25 16:59:43.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/53/805853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/53/805853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/3"}}},"commit":{"parents":[{"commit":"087f89dd16c2fbafaa0aab8e5e2cff09489d40ed","subject":"Add secure-rbac test for Secrets across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/087f89dd16c2fbafaa0aab8e5e2cff09489d40ed"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 13:35:52.000000000","tz":0},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-25 16:58:14.000000000","tz":0},"subject":"Add secure-rbac test for Containers","message":"Add secure-rbac test for Containers\n\nThis patch adds rbac tests to the Containers resource to test\nsecure-rbac policies within a project\n\nThis patch also removes the use of do_request in the existing tests\nas that method is being deprecated in favor of using the clients\ndirectly.\n\n* PEP8 Fixes\n* Fix the plurality of the method names\n* Remove _by_id from certain methods to maintain consistency\n\nChange-Id: I80aba2934110965866d1583309df7f2ca9ef4c27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/34bcf65ad3ddbbc608382f42cf9fb5e67fa0b439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/34bcf65ad3ddbbc608382f42cf9fb5e67fa0b439"}]},"branch":"refs/heads/master"},"4e5e45748d310f7e944dee17427205b430a954d9":{"kind":"REWORK","_number":4,"created":"2021-08-26 18:18:17.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/53/805853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/53/805853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/53/805853/4"}}},"commit":{"parents":[{"commit":"087f89dd16c2fbafaa0aab8e5e2cff09489d40ed","subject":"Add secure-rbac test for Secrets across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/087f89dd16c2fbafaa0aab8e5e2cff09489d40ed"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"cdwilde@gmail.com","date":"2021-08-24 13:35:52.000000000","tz":0},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-08-26 18:17:18.000000000","tz":0},"subject":"Add secure-rbac test for Containers","message":"Add secure-rbac test for Containers\n\nThis patch adds rbac tests to the Containers resource to test\nsecure-rbac policies within a project\n\nThis patch also removes the use of do_request in the existing tests\nas that method is being deprecated in favor of using the clients\ndirectly.\n\n* PEP8 Fixes\n* Fix the plurality of the method names\n* Remove _by_id from certain methods to maintain consistency\n\nChange-Id: I80aba2934110965866d1583309df7f2ca9ef4c27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4e5e45748d310f7e944dee17427205b430a954d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4e5e45748d310f7e944dee17427205b430a954d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
