)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~806982","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Iff17a51d6d23f99376119249c046a84cdfc438e0","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Iff17a51d6d23f99376119249c046a84cdfc438e0","subject":"Add secure-rbac tests for Quotas across projects","status":"MERGED","created":"2021-09-01 20:36:41.000000000","updated":"2021-09-16 19:30:18.000000000","submitted":"2021-09-16 19:30:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"806982-secure-rbac","meta_rev_id":"59f802429e8982802637023aa88e086b2522d413","_number":806982,"virtual_id_number":806982,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 19:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2021-09-15 19:02:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-09-16 18:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-09-16 18:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 21:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 18:31:30.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},"state":"REVIEWER"},{"updated":"2021-09-15 18:31:30.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-09-15 19:02:14.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"7ece6f7d5cdac76cc40f35e5f8881c5a311c9816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-01 20:36:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff2aff54c82eb94a47735825c046d77cca8188a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 21:27:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98e518abedb4e7abc65365e7c79a811 : FAILURE in 6m 16s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ef85a21ee3cc4c64bfd15a6cb36b230c : SUCCESS in 39m 21s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/20817720cb50411da76b1be67fadf5c6 : SUCCESS in 39m 04s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/aa587c8bfdf84fc4bbf85da53491a2da : SUCCESS in 41m 56s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/ceaacbd611e64f31939b7a7bead2e8b3 : SUCCESS in 45m 39s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e6f2e720b2fd4eb09684c106659b762a : SUCCESS in 44m 30s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/81c84445dfc1495f9b315ef9ae6e87c4 : SUCCESS in 42m 10s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/f78f14097ecc4d8984cc02aac6bd435e : SUCCESS in 37m 56s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/d123fbf6a31149a2a0f32d410f3322f5 : SUCCESS in 42m 33s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a220fe5e2104403d8615df2b882b384a : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4888690610501e4abf6742442c8792848a20d890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-02 15:52:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19085ea508e0b4a5abd04a5a4907cbdf03fe0846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:47:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23c06f734034800a74e464c0d4b539a : SUCCESS in 4m 27s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/97cd483e5b20451880077b01606651ef : SUCCESS in 40m 16s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/e66fc6d2886b409f8283192ecad03fe1 : SUCCESS in 34m 18s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/8ef3f3577f58447888d3afe8fbd754a1 : SUCCESS in 44m 13s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/e57db517660f484a909e560676035df5 : SUCCESS in 43m 06s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/c10f3dfbe7af4949b286811993dc7647 : SUCCESS in 36m 51s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/4d0fff32be6945c29e1694e055d53fb1 : SUCCESS in 36m 32s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6c3da3f958a4437da35d3e458f9bf87a : SUCCESS in 44m 49s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/64564c6d5e7740e2ac1ee3bc66a2f2f2 : SUCCESS in 42m 53s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/45d3044b2e14441489c60eceb81c3408 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"49534051f6d427b353c35540cd0e77e68115eafd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-10 18:45:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"aed165a9460ba1e8ba5395278fa252488e07e79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:33:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04441f7e408d42f580ade239c0c452de : SUCCESS in 5m 14s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/758f0271d4f64b86a8950c1945a4f360 : SUCCESS in 37m 27s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/d76baf2ebbd5458082f92a6967c9a6c1 : SUCCESS in 29m 26s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/050c1c3f515d4d5bb93f2e16231f8a00 : SUCCESS in 46m 16s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/792c6a4a45ec4b02ae1369c8e2b29602 : SUCCESS in 36m 13s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/b5b5417176e94560b68f4db971581892 : SUCCESS in 42m 41s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/be917dea576b4b4a9e86483cda9f2a0b : SUCCESS in 35m 46s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b57f8a56395e473a8e7c3007b66d9213 : SUCCESS in 30m 42s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/276cac16eda94a649acf04065fe9f7b0 : SUCCESS in 28m 55s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/82c5b37a3d794dd3818ce9ec83e6b646 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f7a5af5e5e0c74d1e7b502e59a91d86b76a0133e","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-09-15 19:02:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a69f1673d3647d6e8774893a2175b13adb75e8e8","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-16 15:12:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7bfa759abce0f2586fac0f3302fff6afb38611c6","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-16 15:18:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4725c88d44a68f1a374e6f4868a06e257ecba98a","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-16 18:31:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nfine with consistency I guess ..","accounts_in_message":[],"_revision_number":3},{"id":"a06306ee8681319f4ea00131180e308e133c4271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 18:32:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"152e4f6cff308f6c15dbd086199bf94bb60dc2c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:30:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23bf16fd4613457eb29e17464deec626 : SUCCESS in 5m 15s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/138e7097551642a587313a85e865fd9c : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"59f802429e8982802637023aa88e086b2522d413","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:30:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8400a79a94cfa2247dd259414e12b65918178e1e","revisions":{"fd5bb725bdca08d7a74a1ecc511cac250e739f2f":{"kind":"REWORK","_number":1,"created":"2021-09-01 20:36:41.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/82/806982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/82/806982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/1"}}},"commit":{"parents":[{"commit":"ced0ebe88f3d874bd8617ae04bf72bda2947f64e","subject":"Add secure-rbac tests for Orders across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ced0ebe88f3d874bd8617ae04bf72bda2947f64e"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-01 20:35:45.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-01 20:35:45.000000000","tz":0},"subject":"Add secure-rbac tests for Quotas across projects","message":"Add secure-rbac tests for Quotas across projects\n\nThis patch refactors some test for Quotas and adds tests for\nchecking quotas for different projects than the one the persona\nis authenticated as.\n\nChange-Id: Iff17a51d6d23f99376119249c046a84cdfc438e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fd5bb725bdca08d7a74a1ecc511cac250e739f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fd5bb725bdca08d7a74a1ecc511cac250e739f2f"}]},"branch":"refs/heads/master"},"a02e1bcdb3db969199ad2bce74eba0662655535d":{"kind":"REWORK","_number":2,"created":"2021-09-02 15:52:14.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/82/806982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/82/806982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/2"}}},"commit":{"parents":[{"commit":"ced0ebe88f3d874bd8617ae04bf72bda2947f64e","subject":"Add secure-rbac tests for Orders across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ced0ebe88f3d874bd8617ae04bf72bda2947f64e"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-01 20:35:45.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-02 15:51:06.000000000","tz":0},"subject":"Add secure-rbac tests for Quotas across projects","message":"Add secure-rbac tests for Quotas across projects\n\nThis patch refactors some test for Quotas and adds tests for\nchecking quotas for different projects than the one the persona\nis authenticated as.\n\nChange-Id: Iff17a51d6d23f99376119249c046a84cdfc438e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a02e1bcdb3db969199ad2bce74eba0662655535d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a02e1bcdb3db969199ad2bce74eba0662655535d"}]},"branch":"refs/heads/master"},"8400a79a94cfa2247dd259414e12b65918178e1e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-10 18:45:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/82/806982/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/82/806982/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/82/806982/3"}}},"commit":{"parents":[{"commit":"e20bc479827ed28322194b64dc460a1938cf279d","subject":"Enforce scope in RBAC testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/e20bc479827ed28322194b64dc460a1938cf279d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-01 20:35:45.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-10 18:45:25.000000000","tz":0},"subject":"Add secure-rbac tests for Quotas across projects","message":"Add secure-rbac tests for Quotas across projects\n\nThis patch refactors some test for Quotas and adds tests for\nchecking quotas for different projects than the one the persona\nis authenticated as.\n\nChange-Id: Iff17a51d6d23f99376119249c046a84cdfc438e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8400a79a94cfa2247dd259414e12b65918178e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8400a79a94cfa2247dd259414e12b65918178e1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
