)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~812178","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I03c1126c416f7c3aacce431bab99606b4a80eccd","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"test_branch","hashtags":[],"change_id":"I03c1126c416f7c3aacce431bab99606b4a80eccd","subject":"Testing - DNM","status":"ABANDONED","created":"2021-10-01 17:15:27.000000000","updated":"2021-11-10 21:42:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2824951482a35ef0de0eb7c367c8c4da695db8c5","_number":812178,"virtual_id_number":812178,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-01 17:31:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-01 17:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a33f4955064dcf6e32bf0ec5ea670dfd6d84576f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-01 17:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a09c5c117f5ca71ba6a53ee339c067ce454f612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:31:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/179cefa0d76c44eb8ef8298d89fbb8d3 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2094d740037e4558b10d78a8977a9956 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2824951482a35ef0de0eb7c367c8c4da695db8c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-10 21:42:03.000000000","message":"Abandoned\n\nAbandoning DNM for lack of activity.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d648c7a92a6463c66d974a9e32b00ce82ce29368","revisions":{"d648c7a92a6463c66d974a9e32b00ce82ce29368":{"kind":"REWORK","_number":1,"created":"2021-10-01 17:15:27.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/78/812178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/78/812178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/78/812178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/78/812178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/78/812178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/78/812178/1"}}},"commit":{"parents":[{"commit":"8400a79a94cfa2247dd259414e12b65918178e1e","subject":"Add secure-rbac tests for Quotas across projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8400a79a94cfa2247dd259414e12b65918178e1e"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-01 17:14:46.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-01 17:14:46.000000000","tz":-240},"subject":"Testing - DNM","message":"Testing - DNM\n\nChange-Id: I03c1126c416f7c3aacce431bab99606b4a80eccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d648c7a92a6463c66d974a9e32b00ce82ce29368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d648c7a92a6463c66d974a9e32b00ce82ce29368"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
