)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~814229","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I4e01c0e7f93f52c72faadb9d2a8317f9f553904b","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2021-11-04 03:43:58.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2021-10-18 20:31:46.000000000","reason":"David Wilde replied on the change"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2021-10-27 13:56:36.000000000","reason":"Ade Lee replied on the change"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2021-11-04 03:43:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e01c0e7f93f52c72faadb9d2a8317f9f553904b","subject":"Add secure-rbac test for Container ACL API","status":"MERGED","created":"2021-10-15 20:08:26.000000000","updated":"2021-11-04 03:43:58.000000000","submitted":"2021-11-04 03:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814229-secure-rbac","meta_rev_id":"c5df58322f032147eafa845992f1a9d9e00bae84","_number":814229,"virtual_id_number":814229,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 03:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2021-11-04 03:43:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-11-04 03:43:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-11-04 02:56:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 20:09:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-10-15 20:09:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-10-15 20:09:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2021-10-15 20:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ca71a30eb732cd33c187b386962ec3797a20953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-10-15 20:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06748c223d63024eebd867aacf62ab85f0033521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 20:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcab7cc8212847eb8e9b88f70deb8221 : SUCCESS in 3m 50s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9194d28a8100494f9cea1129aa51486e : SUCCESS in 34m 39s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/24fb6de4d30341a39dc2f23f1ada396e : SUCCESS in 31m 25s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/b1e64b326de343709ea6e747b2ce977c : SUCCESS in 44m 22s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/bbdf2a7ba8bc43f49362259ee8797c9c : SUCCESS in 38m 12s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/6a9b2e9aa8fa46afa8667bd5fdefe0d2 : SUCCESS in 39m 32s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/72ff519431984a5ab1cdb030ada7a687 : SUCCESS in 37m 38s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/3054057d324a4128a7c55f5424539d30 : SUCCESS in 34m 49s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/8ebe979065224c68b2fcb4279fc2a60a : SUCCESS in 33m 15s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e3455335b01e4b2c8d8cdc140a1675a4 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4bf3c09169641b5760f6b5e576c4c72a65dbf222","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-10-18 20:31:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fbf2c2eac34eb1cc56256f64878874c77e1e633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-10-27 13:53:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9c4fb33fb7d415664259ca9242822ac08ea3325c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-27 13:56:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5907c75833f98efc0b39d3738c4c6629caaddc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 14:55:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97998d1dadb942baaeee0495dc521050 : SUCCESS in 4m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ad4a5f1ee91447859cb81dfab6d195fe : SUCCESS in 37m 27s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c3b7fa5d88c74750a6a4976d47c51208 : SUCCESS in 35m 51s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/40db12706f0847f8835302b633d8fbbf : SUCCESS in 44m 56s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/78d0a213d28a4adc9456cf2710d76e75 : FAILURE in 43m 29s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e3fda984cc034aedb0b199eeeaa98392 : SUCCESS in 40m 28s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/652fd7519f244ae08ebca3b82bcd0fb9 : SUCCESS in 53m 15s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c482f14814ce43b1ac9560d93b5ced0b : SUCCESS in 38m 17s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/250f006d892d4dc8925184cebabef6d1 : SUCCESS in 44m 35s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a16f1220f0c64a318f1ae9e7379622cb : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9b983cfc837d7b4c6bd3f8f29d428b40578ae452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-03 17:00:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e663c186649eeb300ba4dcc82fdcf0c3c6e55acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 18:03:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba5d7bf916340d6ad7481a882fb28dc : SUCCESS in 4m 44s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/937dcab578e54a95a65377a9f23977fa : SUCCESS in 45m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/d51835a55d264165a6aace48d2bc4230 : SUCCESS in 37m 01s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/35c3d1ee99e043ecbda40006e7ec2320 : SUCCESS in 55m 05s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/a8e7871f1e9442d3bbeead0086e957a3 : SUCCESS in 47m 43s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f01c605acf9347aaa55320fa7b3c9584 : SUCCESS in 47m 25s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/e2a15df10cc143a3a50ffb33fc0db9c8 : SUCCESS in 51m 41s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/13768ec5f26245c2a2b4da27c290721e : SUCCESS in 39m 41s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/e5cf24771ce04db7948386bf003b30dc : SUCCESS in 40m 35s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/6b0a861c758b45e68070b76af8749f61 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f6bb9a9c90249a9523d7127a5b7675f74a8503ca","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-04 02:56:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a6a12bb120bb4c0e115e3970282d56693f381965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 02:57:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e14185fb69a99fac76fe1bc104d1e3eb8b4d0407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 03:43:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ddd305c70448df83ecd27c5821b586 : SUCCESS in 4m 55s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2e771b0c64a544d08ef3c5a29e94ec8f : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0c3bacfec88b2e94fbe56a7883a6435cc5c5aa9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 03:43:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"378efe61acd169f4edb5601df8d4cf972acd4813","revisions":{"a5f23f7ce97a5a6b16fefc70bba75256fe98471f":{"kind":"REWORK","_number":1,"created":"2021-10-15 20:08:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/814229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/29/814229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/1"}}},"commit":{"parents":[{"commit":"93361b4eba489a1079d6ee7873e098081f8bbc99","subject":"Add cross-project tests for Containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/93361b4eba489a1079d6ee7873e098081f8bbc99"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-15 20:07:52.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-15 20:07:52.000000000","tz":0},"subject":"Add secure-rbac test for Container ACL API","message":"Add secure-rbac test for Container ACL API\n\nThis patch adds policy tests for the Containers ACL API.\n\nChange-Id: I4e01c0e7f93f52c72faadb9d2a8317f9f553904b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a5f23f7ce97a5a6b16fefc70bba75256fe98471f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a5f23f7ce97a5a6b16fefc70bba75256fe98471f"}]},"branch":"refs/heads/master"},"f0f193607606b9c7bfdad91f1c985bd2738c9888":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-27 13:53:38.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/814229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/29/814229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/2"}}},"commit":{"parents":[{"commit":"70924f4ab1258c4fa30bda477951ede2e0d24720","subject":"Add cross-project tests for Containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/70924f4ab1258c4fa30bda477951ede2e0d24720"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-15 20:07:52.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-10-27 13:53:38.000000000","tz":0},"subject":"Add secure-rbac test for Container ACL API","message":"Add secure-rbac test for Container ACL API\n\nThis patch adds policy tests for the Containers ACL API.\n\nChange-Id: I4e01c0e7f93f52c72faadb9d2a8317f9f553904b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f0f193607606b9c7bfdad91f1c985bd2738c9888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f0f193607606b9c7bfdad91f1c985bd2738c9888"}]},"branch":"refs/heads/master","description":"Rebase"},"378efe61acd169f4edb5601df8d4cf972acd4813":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-03 17:00:22.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/814229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/29/814229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/29/814229/3"}}},"commit":{"parents":[{"commit":"e80c1db7e022e6d60548a25488c1f1147f392eb1","subject":"Add cross-project tests for Containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/e80c1db7e022e6d60548a25488c1f1147f392eb1"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-15 20:07:52.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-03 17:00:22.000000000","tz":0},"subject":"Add secure-rbac test for Container ACL API","message":"Add secure-rbac test for Container ACL API\n\nThis patch adds policy tests for the Containers ACL API.\n\nChange-Id: I4e01c0e7f93f52c72faadb9d2a8317f9f553904b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/378efe61acd169f4edb5601df8d4cf972acd4813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/378efe61acd169f4edb5601df8d4cf972acd4813"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
