)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~815071","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~If2209b12dce107c5648d39270d977a1e9f3bea1d","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-02-11 16:50:26.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-02-11 16:50:26.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-02-07 22:08:58.000000000","reason":"Ade Lee replied on the change"}},"hashtags":[],"change_id":"If2209b12dce107c5648d39270d977a1e9f3bea1d","subject":"Add secure-rbac test for Consumers","status":"MERGED","created":"2021-10-21 20:50:36.000000000","updated":"2022-02-11 16:50:26.000000000","submitted":"2022-02-11 16:50:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815071-secure-rbac","meta_rev_id":"3f26e1ed39b80643477c2e06a131e1093063b11a","_number":815071,"virtual_id_number":815071,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 16:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-02-11 16:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-02-11 16:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 22:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 13:12:37.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2021-11-12 13:12:37.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"15e0aced523e35ec614c73723646619547ff47b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-10-21 20:50:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34316a6fb4d109c1f0dfa84bede40f9e3a1ae8d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:00:46.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5867b4fefb784ce9afb6acbd90123730 : FAILURE in 4m 16s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/db34c885436c43529589d643579c8416 : SUCCESS in 43m 54s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/6f911ce48fa440939385e7ac3cbf89aa : FAILURE in 40m 58s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/c8f235320f4647c4bb11644caf267494 : SUCCESS in 54m 53s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/7ac276e1649741d391996f688a651a54 : SUCCESS in 1h 04m 05s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e13fc55f33684f7bafedbaad067f9c2c : SUCCESS in 44m 27s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/0df7750e72144ab3b1ada7c697eaf0bb : SUCCESS in 53m 55s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/7071d3d4a71e4bf08a592a63761ec640 : SUCCESS in 45m 06s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6cbf35e2df82431481c236350e1bf5d6 : SUCCESS in 48m 48s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e2002bf2ebd8468b9cc8c100df388738 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e5a36cac2f26203777dce923f7b34de6bab65940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-03 17:08:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b44df9bceedc2e79fc42b4b2768ebf10da8ac0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 18:03:17.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194b903665194832a4eb78a1777dee72 : FAILURE in 4m 51s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ff670755fe7941e9a36ae041af60a2f7 : SUCCESS in 29m 13s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f782d06c242f470cb3e31c480aeb668e : FAILURE in 40m 09s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/736d719656c547eda9c4d4c411a01d6a : SUCCESS in 48m 31s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/be3f92851b164a018d0ef5027182826f : SUCCESS in 32m 37s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/258a5d1cb7d74a4db20fd8c54230a3b1 : SUCCESS in 33m 24s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/fae023518c4c45b09e1f68d0ebad90d5 : SUCCESS in 48m 51s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/84b05aebf5f3413b816fa4afc3568b41 : SUCCESS in 29m 52s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f499ea77d3c5408f9a8cc24988355e38 : SUCCESS in 45m 18s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f948d7ea02ad42e2bd1b56452e921e41 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4a122c27746f8da46ffe5b9e8aca8c381c870649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-04 22:01:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7b26c0cbc6523affecb10f2ada82914edb8eb576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 22:53:17.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea939da75f64aa980fd5f98c23bcc45 : FAILURE in 4m 05s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c3d9319b0580456297229824e751dc34 : SUCCESS in 36m 45s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4af4625f49c141c8a1c86315bd2c1584 : SUCCESS in 35m 37s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/ea402b55da8a49f4a25ef02949a13944 : SUCCESS in 37m 48s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/40e6d9a0b652455fbd579eef63811b7e : SUCCESS in 42m 02s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/e72f6314d7df4a9c84eab541ac37840c : SUCCESS in 40m 00s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/0972056210bd4d3299f8b83790a4d481 : SUCCESS in 40m 02s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/72da73c082a34cc2adec48852aeb5b7c : SUCCESS in 44m 41s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/4a1768e1f54a476299452e400f1c68bd : SUCCESS in 35m 34s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ec203ebe4df64926829d2b06e57164f6 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d65131b6389f16d4d6fa338b852b20861d68e048","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-05 17:13:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f669c263a92bfc37435b983c8f796eb946d2a06d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 18:00:58.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c928cd382f9441d38dbcea08540073ec : FAILURE in 4m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7062b0ef7a3a4a6c94dc5a9385db0999 : SUCCESS in 35m 03s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4d0305f44c6e4489af45d6b68a56839c : SUCCESS in 27m 06s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/ef59d0f53b604ac48d0f2e1265a2f1e1 : SUCCESS in 33m 20s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/3a70e4b01afd47738f03c9caaf41257d : SUCCESS in 44m 37s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/6c96584f5d5948fcb1de665626ab2be7 : SUCCESS in 35m 47s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/460c64b3f1594ac89e2bd3ea31960e16 : SUCCESS in 40m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a9d176bd444be2b1e4ce664d01636a : SUCCESS in 37m 26s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b597c27c0933453197d692cb73dc346e : SUCCESS in 32m 58s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/13b2411166184fa1871dd03793d3ba71 : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":3},{"id":"85b962434d29ce6538faea10dc22aaa8d617d1f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-11 21:16:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"89000a75555521a1e38325a896b9fa474f65d60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-11 21:26:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e16b757190193732630011df76a69740a0dfee1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 21:36:28.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"14db59d19fed464a473e5a6f15a3ca299819b0e0","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-12 00:55:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"33866abdc19777dd803dc2ceaf8872a7d01fa8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 02:01:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754359bd56844a8c9cc117b1335ceba4 : SUCCESS in 3m 44s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d50529b633314f77ab1c1fb50f491b7a : SUCCESS in 36m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/227d313edc2944d98629683fbb697124 : SUCCESS in 31m 34s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/82fbc6be9e07492f8991e23c50644985 : SUCCESS in 40m 07s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/59025b6617df46469a77e10ef65cd134 : SUCCESS in 37m 48s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/c6f1b6f682204d70832330dcb902b5f0 : SUCCESS in 35m 05s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/a460e4fe34f14976909478b2ad232d0e : SUCCESS in 37m 46s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/fedfcc6f17304542b54354ff0e0bb688 : SUCCESS in 47m 29s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/056851ed23354887bc6d6a25cd48061d : SUCCESS in 35m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/36cfc77d8a5644ccbd2a0830a5c283c8 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":5},{"id":"106ee3f8bfed49bf669149745f5e52862b731802","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-12 21:11:31.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9bf915d7bd5e3eaa548682e0c5a79ded80f51ecc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-29 21:32:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f78919e71d5d2549633d3e7c0f211e4e556024a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-11-29 21:33:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"60bfe66ebd2c150e789c75d442abb131fd97b703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 22:30:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35c749b64b94258bd22556a4254815c : SUCCESS in 3m 11s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/97d66554b6c84ad1bf5b35779a2ce3aa : SUCCESS in 42m 06s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/514e6c9e5f934a1194a87f0614c19644 : SUCCESS in 34m 22s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/2322f5e048bf4b94af063229713b5074 : SUCCESS in 49m 25s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/deaf6dc14968453bbcd1ade7ae804cfd : SUCCESS in 41m 22s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/da12f7e5c4e64b09ab68ba4c8613b338 : SUCCESS in 47m 53s\n- barbican-tempest-plugin-simple-crypto-train https://zuul.opendev.org/t/openstack/build/2201532990744cfba23f02dc543e014e : SUCCESS in 35m 42s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c9401caa0db8468795b89f04aef97b91 : SUCCESS in 49m 25s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/d5717c0e26c44896bc77c9ab2484c02f : SUCCESS in 41m 06s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/80644472184b48d7b04a74e7d8f42f8e : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":6},{"id":"91764bcbe7e31028ff33d988f5a0d6f385aaa7ad","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-01-31 19:49:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9773bb31ff97a335e5c7e2bf6c0279192c02f576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:43:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f571f2992d4e46a9acbdfbc3861382ad : SUCCESS in 3m 32s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/aa203bc5eaf8432c9366088245225fa3 : SUCCESS in 39m 43s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/deb999e95b004c5ab2b4e68bf190bc96 : SUCCESS in 44m 33s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/d16a5e44971a4a0da9ede6596c7b620e : SUCCESS in 38m 22s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/667ea78a3c50424c94e3542f568a0293 : SUCCESS in 41m 01s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/aa19c7b4f72a4cd7ba1304a74b1a4e84 : FAILURE in 14m 31s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/d8caeb3215d4422fb46677386811b5d6 : FAILURE in 10m 46s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/141832bea5bc4b419975ca63c4137492 : SUCCESS in 43m 11s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/3e09b23d730b456eae026a40b931af69 : SUCCESS in 40m 44s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/08843e04a8b24961905f7eae20ea934c : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":6},{"id":"0fe541d61971ea55f5ee654f2ff05760d1f0d320","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-07 22:08:58.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"850045b2547621f6ea9cf838f09126ba55fe5065","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-02-11 14:28:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"542202f5207b90f531843c786eed4191cfe3a720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:20:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c806678a415842a5b7a9f353fb348801 : SUCCESS in 3m 52s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f1dc689cde1c4fb182d13ace37bf4ff8 : SUCCESS in 38m 32s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4a9633d078404089aa1e71fc2b353978 : SUCCESS in 36m 58s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/8033d7fd77ea4fce901cb3cbfdcb5181 : SUCCESS in 38m 22s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/0cfee98263e844fb9504c873e281f223 : SUCCESS in 44m 13s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/439f325f4c4d47f49c96b08df906d0a9 : SUCCESS in 42m 57s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/9f28ca88b8564242816f97db281f4033 : SUCCESS in 44m 10s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/24caf20e5ca848ea99b05ac255ff6d9c : SUCCESS in 45m 31s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/61eecf30ab3d43018e3d5e7d9eabcb3e : SUCCESS in 41m 33s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ed93be02a6d4493abe91ed18661efe5f : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":6},{"id":"50fb18999c664d8b2161242ccc13853d7e7c8085","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-11 16:16:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"feaf648f54b21e75ecef61b8df4096e9799a6a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:17:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e677f73b2b4b8a8efecfcb2bb945c7c5e4fa141e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:50:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bf67a7c4234f6da71cd6960ba1746b : SUCCESS in 3m 45s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b692e7fe2a904e9689d5124307be9955 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":6},{"id":"3f26e1ed39b80643477c2e06a131e1093063b11a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:50:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"82fa4a08a3a7da9b6159b9ffd545976fdce59e63","revisions":{"5af932a60f20e9047a52e26103e5cdef34725b52":{"kind":"REWORK","_number":1,"created":"2021-10-21 20:50:36.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/1"}}},"commit":{"parents":[{"commit":"a5f23f7ce97a5a6b16fefc70bba75256fe98471f","subject":"Add secure-rbac test for Container ACL API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a5f23f7ce97a5a6b16fefc70bba75256fe98471f"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"subject":"WIP - Add secure-rbac test for Consumers","message":"WIP - Add secure-rbac test for Consumers\n\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5af932a60f20e9047a52e26103e5cdef34725b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5af932a60f20e9047a52e26103e5cdef34725b52"}]},"branch":"refs/heads/master"},"9921aa0c5ec2543e9d9f56b69049006952875c21":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-03 17:08:54.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/2"}}},"commit":{"parents":[{"commit":"378efe61acd169f4edb5601df8d4cf972acd4813","subject":"Add secure-rbac test for Container ACL API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/378efe61acd169f4edb5601df8d4cf972acd4813"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-03 17:08:54.000000000","tz":0},"subject":"WIP - Add secure-rbac test for Consumers","message":"WIP - Add secure-rbac test for Consumers\n\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9921aa0c5ec2543e9d9f56b69049006952875c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9921aa0c5ec2543e9d9f56b69049006952875c21"}]},"branch":"refs/heads/master","description":"Rebase"},"d5ae005e3279d7b983b7a5af1153e630e1dc5dfa":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-04 22:01:50.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/3"}}},"commit":{"parents":[{"commit":"378efe61acd169f4edb5601df8d4cf972acd4813","subject":"Add secure-rbac test for Container ACL API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/378efe61acd169f4edb5601df8d4cf972acd4813"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-04 22:01:49.000000000","tz":0},"subject":"WIP - Add secure-rbac test for Consumers","message":"WIP - Add secure-rbac test for Consumers\n\nDepends-On: I1724152839f0f5850f8d32d40b36d1670c0ad996\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d5ae005e3279d7b983b7a5af1153e630e1dc5dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d5ae005e3279d7b983b7a5af1153e630e1dc5dfa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"522d84d80ac1304955ea53330ae53901769cef61":{"kind":"REWORK","_number":4,"created":"2021-11-11 21:16:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/4"}}},"commit":{"parents":[{"commit":"263a326d05663692427870a665e0615c512e4b76","subject":"Temporarily delete two bad tests for consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/263a326d05663692427870a665e0615c512e4b76"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-11 21:15:37.000000000","tz":-360},"subject":"Add secure-rbac test for Consumers","message":"Add secure-rbac test for Consumers\n\nThis patch adds tests for Container Consumers.\n\nDepends-On: I1724152839f0f5850f8d32d40b36d1670c0ad996\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/522d84d80ac1304955ea53330ae53901769cef61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/522d84d80ac1304955ea53330ae53901769cef61"}]},"branch":"refs/heads/master"},"195471cb3d53be712f6f1e39d0b9d6bf6ca906b7":{"kind":"REWORK","_number":5,"created":"2021-11-11 21:26:40.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/5"}}},"commit":{"parents":[{"commit":"263a326d05663692427870a665e0615c512e4b76","subject":"Temporarily delete two bad tests for consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/263a326d05663692427870a665e0615c512e4b76"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-11 21:26:27.000000000","tz":-360},"subject":"Add secure-rbac test for Consumers","message":"Add secure-rbac test for Consumers\n\nThis patch adds tests for Container Consumers.\n\nDepends-On: I1724152839f0f5850f8d32d40b36d1670c0ad996\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/195471cb3d53be712f6f1e39d0b9d6bf6ca906b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/195471cb3d53be712f6f1e39d0b9d6bf6ca906b7"}]},"branch":"refs/heads/master"},"82fa4a08a3a7da9b6159b9ffd545976fdce59e63":{"kind":"REWORK","_number":6,"created":"2021-11-29 21:33:33.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/71/815071/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/71/815071/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/71/815071/6"}}},"commit":{"parents":[{"commit":"263a326d05663692427870a665e0615c512e4b76","subject":"Temporarily delete two bad tests for consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/263a326d05663692427870a665e0615c512e4b76"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2021-10-21 20:50:27.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-11-29 21:33:02.000000000","tz":-360},"subject":"Add secure-rbac test for Consumers","message":"Add secure-rbac test for Consumers\n\nThis patch adds tests for Container Consumers.\n\nDepends-On: I1724152839f0f5850f8d32d40b36d1670c0ad996\nChange-Id: If2209b12dce107c5648d39270d977a1e9f3bea1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82fa4a08a3a7da9b6159b9ffd545976fdce59e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82fa4a08a3a7da9b6159b9ffd545976fdce59e63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
