)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~829207","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I38616a6be43619abf7e5d51e4884722dec4369a6","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-04-18 07:31:30.000000000","reason":"Change was marked work in progress"},"30742":{"account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"last_update":"2022-04-18 07:31:30.000000000","reason":"Change was marked work in progress"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-04-18 07:31:30.000000000","reason":"Change was marked work in progress"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-04-18 07:31:30.000000000","reason":"Change was marked work in progress"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-04-13 07:50:18.000000000","reason":"Martin Kopec replied on the change"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-03-15 21:25:42.000000000","reason":"Ade Lee replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-04-18 07:31:30.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I38616a6be43619abf7e5d51e4884722dec4369a6","subject":"Remove setup_clients() api","status":"NEW","created":"2022-02-15 09:55:08.000000000","updated":"2022-04-18 07:31:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":8,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ac5e0a0fa000042deafa0eb344f99b4cb21db396","_number":829207,"virtual_id_number":829207,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-12 11:56:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":-1,"date":"2022-04-13 07:50:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 10:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 07:38:02.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-03-15 07:38:02.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-03-15 07:38:02.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-03-15 07:38:02.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2022-03-15 21:25:42.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-03-15 21:25:42.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-04-13 09:06:03.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"state":"CC"}],"messages":[{"id":"c4ba2d44b4b46b2b4699e022595cc527e6deaf31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-15 09:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f76d00f0c0a3a220eb0b917c44fca34c0dee072c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-15 09:55:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"377a1dd5e7078199d1adfa5f38b75a4ac62744df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:41:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a069cfa197f4323b837ce1c8a575cbe : SUCCESS in 3m 18s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6628c6647f59415c8a035a64e8992f25 : FAILURE in 33m 53s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/48502b9562b948d188b407bbd3945bd7 : SUCCESS in 38m 15s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/fb53aacf4679442e901d617e866583e3 : FAILURE in 39m 27s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/dd25448e4d4f45b0b72bf2de49e79813 : FAILURE in 43m 35s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/5fad9638f4b4441c8a75906ca9936616 : FAILURE in 45m 04s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/4b59ca142777418cac20ecabe48fe6a3 : FAILURE in 41m 04s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/e64f7a0973404bafa4b1951174f5dd99 : FAILURE in 38m 08s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/35bf72961ec74c1e896df557a2b402d0 : FAILURE in 34m 11s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/7a3d60fcf70a462ba54f4e5c1333e56d : FAILURE in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e4933d79e63bc5d2dcbe7c6adb1411fd1a5d5b33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-16 11:37:21.000000000","message":"Topic barbican-stable-scenario-manager-consumption removed","accounts_in_message":[],"_revision_number":1},{"id":"b540669649b5bb3977ccff4804e5aca0b1044843","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-16 11:37:36.000000000","message":"Topic set to tempest-scenario-manager-cleanup","accounts_in_message":[],"_revision_number":1},{"id":"5d10cd32731f73588641aaaf6ac9f0aeba1f0333","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-28 14:13:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"53296345356a5395ab0ad6b253326c03d2c7b838","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-02-28 14:13:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e8ea3951cb4ca9d24086b990cde40aa96d4e157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 15:01:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fcb878c3254b9e92010bca3f3dde77 : SUCCESS in 5m 03s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0bedfb77b2d54eea9d863fa151ae19f2 : FAILURE in 33m 39s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/1a16296907684dcd87b873208485360c : SUCCESS in 34m 47s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/42f61fcddaeb43d6a0581fa8f8220683 : FAILURE in 36m 16s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/c65112b01800489487cd4d73def093a7 : FAILURE in 44m 57s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/0c5bbe183d0b4d75abdd9ec54329f073 : FAILURE in 40m 51s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f8fa6e26003e4171aa8d9343205ed57e : FAILURE in 38m 08s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/18d9d769230a47fab84177463cbada44 : FAILURE in 36m 24s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/8d098570fb0e47a39b99c94ee83362c6 : FAILURE in 41m 11s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ddaa0381a0ec4e3a8909c05d29485cd4 : FAILURE in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c55b388f77f2efa553616804c6fe65d170ce95c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-01 11:51:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"46589390b0383c74fa38762bc297dacf6ddf005e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-01 11:55:53.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ae0b9848a7ba3ca14ec8329f91a8f46c3814a9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:56:31.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/barbican-tempest-plugin for 829207,3","accounts_in_message":[],"_revision_number":3},{"id":"002439c38870953a15f745408e5449cb6efe6747","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-01 11:57:41.000000000","message":"Topic barbican-stable-scenario-manager-consumption removed","accounts_in_message":[],"_revision_number":3},{"id":"bd81d8e5d700ee3432306eab5105e6fcc109ae06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-01 11:57:52.000000000","message":"Topic set to tempest-scenario-manager-cleanup","accounts_in_message":[],"_revision_number":3},{"id":"36a19f1d2352cb2d6d875ea3f1489e5982f1ba47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-15 07:37:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"751c3bdf36aa38a1796b65c936fd23bbdeaa4300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 08:33:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c02ebda0dd4f798d43244c65b0d048 : SUCCESS in 4m 34s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/379d825aa84f48d68055135bb25e6918 : FAILURE in 41m 28s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/5e3fa6e17267425999453cf35c887714 : SUCCESS in 43m 34s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/257d11111bd8498795f68057adf6219d : FAILURE in 43m 46s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/357ac14906d34d6ca5649b1390004e67 : FAILURE in 48m 04s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/b4a77e5f502f42d2be548ce972d78bfe : FAILURE in 53m 38s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/5b4cd92f70d94ad7bf18fe90d90e8552 : FAILURE in 45m 09s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ada181f4be494b68964fbacf2f48fe87 : FAILURE in 45m 14s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ef2bafd390124ab2ab89860bd4a4b94f : FAILURE in 41m 28s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/d1a0d2aef38c46bc8fddfb40865a5155 : FAILURE in 34m 14s","accounts_in_message":[],"_revision_number":4},{"id":"a62ad94fb86debefe2ab148fd04d0cf98053aab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-15 10:22:01.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ea444cc2aa931b73c8e4fff07e72da139ac39a41","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-15 10:44:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff74f938a21ace45be660f4b2ca5ef5e86cd6cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 11:27:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c761d60c494b73a12850ba47b8d53a : SUCCESS in 5m 25s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6dd0eb4af6204db4a1f2b46ff10975a6 : FAILURE in 31m 18s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4924ac0627a0425d90d57af284392351 : SUCCESS in 29m 09s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/43a3f69feddb4736b59ba79a2aefe372 : FAILURE in 28m 41s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/4369a7b056974cd3a61d122cfe7e41c0 : FAILURE in 49m 13s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/697d89233a3449c6b8593e5ca0c55eaa : FAILURE in 52m 25s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/47e7ec84450f4c16b6dfb4093e8eb313 : FAILURE in 45m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/17aa764029f34a5eac694d5a96f46d3e : FAILURE in 49m 31s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/83b8dba1cabd4bfdb3600fd970f636b8 : FAILURE in 49m 25s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/cb14288800754b20bf24c14b383d3402 : FAILURE in 49m 27s","accounts_in_message":[],"_revision_number":5},{"id":"ed3bd072a6046ad2ae90660cf0a385df34d80c24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-15 13:31:51.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"bbf5f3868712229c9b0cef61bf938b7a4373cb30","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-15 13:32:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f4a3da5709073f0ce07c2374ecf5fab3be2967d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:32:48.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tempest for 833788,1","accounts_in_message":[],"_revision_number":6},{"id":"1780a23d92f9aaaf924b8dce541ac101f3556099","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-15 20:53:33.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"122fce0a89a490b3a86ef2ab8522380b2c1eb78c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 12:13:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"21e9049b7374e3a2296468805c265eea8895eb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 12:14:21.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tempest for 833758,2","accounts_in_message":[],"_revision_number":7},{"id":"af96ad6f4457412f710ff390dc65e1aea2d8ee54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 13:24:49.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"48ddd7f2542785364c0ac5491ad6a1d09de24082","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 13:25:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"79cce288d92bf16e7e93a91204ae914d2dcb4b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 14:08:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24e73caf1b24b5c8e9c1dc0a1662bba : SUCCESS in 5m 44s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/32aa385c880546b1b68c725fd456cc33 : FAILURE in 36m 23s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/49f25e31f6f44048885495249128550f : SUCCESS in 34m 30s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/04df0f85e3524d53afb65e2dcd2f7663 : FAILURE in 34m 42s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/9f162c73538f49b5a4038b4515994fd5 : FAILURE in 38m 30s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/993861ac75364448a449f52b744116ce : FAILURE in 37m 48s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/b9c54d78de004186a8024f82a5367af4 : FAILURE in 37m 32s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/07015803bc2b459faab8441d9724d408 : FAILURE in 35m 08s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/91c0ca29f0d747aa83adaaca0a977ae2 : FAILURE in 39m 06s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/fcf813d4bd71436f8564e1eea1c30cbd : FAILURE in 36m 43s","accounts_in_message":[],"_revision_number":8},{"id":"81020e0469f8c442c138622408b60183da96955a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 17:58:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b06360877d6e55b17f3c23a076b65ffe2d346196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 18:15:17.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"264e5e7fbf06009736cb3cf39f10518b21958198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:04:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd6b910c2c8469da6c0939f3b18abc9 : SUCCESS in 4m 37s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/da0e013592da478db4503edf8f684c27 : FAILURE in 40m 00s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/87b1f83454ad4e45889120fb1112dcb9 : SUCCESS in 39m 29s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/86a17ca3e3e141f18bad9599449f7eb2 : FAILURE in 37m 20s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/5a6e40103847483f9988c01e5ef36335 : FAILURE in 43m 01s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/57384f127fb04efb93a73da95857514b : FAILURE in 42m 38s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/52c853f17f3546b4b74becf7d36c3dd0 : FAILURE in 43m 36s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6118a5768ca543cd91de72f5ffbd9000 : FAILURE in 41m 06s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/eec9e62de79649488644d942433407f0 : FAILURE in 41m 56s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/d8b2f8906d6744e2a5b852cd2a18db8e : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":10},{"id":"2eb7d99da8b72c6db7b61930a852da8c7eb85384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-04-12 10:40:21.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"9017f093061338d5e4ec80a70a8bafd8ab3d4d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 11:56:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86875c3bc2fc438ab649a69837d64ef3 : SUCCESS in 3m 41s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c4a8d8ed842c47d496f179b9ca1fe9ea : FAILURE in 40m 36s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/8eef819f675e4bdd978bdc09a0c6f1ca : SUCCESS in 47m 59s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/ce050d5114b548fca4e51690077dd198 : FAILURE in 52m 19s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/5b192c5910ce47a2beb29ca5b20edf15 : FAILURE in 37m 57s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/478ac3c4c0b24ee7bad201d5ea96c7da : FAILURE in 41m 11s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/f4b9f865b94e4e7997eb73c88e5840f9 : FAILURE in 37m 39s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d1c9b7b3434a8ba9eb4ae90beb38d0 : FAILURE in 34m 58s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/624020568c1f4bbaa6bccda8d83550ea : FAILURE in 31m 31s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/b8228f8aa45a4b119a9ff4972b86f9d8 : FAILURE in 44m 02s","accounts_in_message":[],"_revision_number":11},{"id":"397760f4e8ed25d185c5d95faa1cd8cab63b8f51","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-04-12 13:47:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7616925c1a90843ecf83661791d74914c6960e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:50:53.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbbd9d3d7134d658a401366a9b1f9a2 : SUCCESS in 3m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fa9631dc0c784008bc7a9e55caea4645 : FAILURE in 42m 25s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/980430818cdd457087987950844996bc : SUCCESS in 40m 15s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/1c0d81405be74d37ab1777582e2b6853 : FAILURE in 37m 31s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/f65c1e665ba542eb9db34671d591b0d5 : FAILURE in 43m 51s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/c8712a4ca26942ac9798f16dc21e8c39 : FAILURE in 42m 12s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/506b046b4e3144dfbae255b36d3170fc : FAILURE in 43m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/043c04252420462eb05f8fd1504217f6 : FAILURE in 42m 38s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/9786b1ede05e4460a5bf1a6a8a0a1c20 : FAILURE in 52m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/184c8b4ab97c467a91926080ffaa24a4 : FAILURE in 42m 33s","accounts_in_message":[],"_revision_number":11},{"id":"410ad894f71e199b1001a0c1cdd9a67bcb931900","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-13 07:50:18.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c9be1a63bf2e5b46e47769e6bb0364c08069c0ee","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-13 09:06:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ac5e0a0fa000042deafa0eb344f99b4cb21db396","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-04-18 07:31:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"315fe69cc24ea9705d8ab39cca6e417dd958978a","revisions":{"59c95b0dd50eb923bbfc02f2f46a44c99c1ee5cb":{"kind":"REWORK","_number":1,"created":"2022-02-15 09:55:08.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/1"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-02-15 09:01:56.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-02-15 09:54:43.000000000","tz":330},"subject":"Remove the copy of scenario manager","message":"Remove the copy of scenario manager\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/59c95b0dd50eb923bbfc02f2f46a44c99c1ee5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/59c95b0dd50eb923bbfc02f2f46a44c99c1ee5cb"}]},"branch":"refs/heads/master"},"654a89122bb17573d6c752583a8a03c330a90861":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-01 11:51:40.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/2"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-02-15 09:01:56.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-01 11:49:24.000000000","tz":330},"subject":"Remove the copy of scenario manager","message":"Remove the copy of scenario manager\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n2. create_volume()\n3. create_volume_type()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/654a89122bb17573d6c752583a8a03c330a90861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/654a89122bb17573d6c752583a8a03c330a90861"}]},"branch":"refs/heads/master"},"89ba4c549bd628bc6c213302dd02250f904b0343":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-01 11:55:53.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/3"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-02-15 09:01:56.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-01 11:54:34.000000000","tz":330},"subject":"Remove the copy of scenario manager","message":"Remove the copy of scenario manager\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n2. create_volume()\n3. create_volume_type()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\nDepends-On: I3f9209fef4da9cc85d78060da0bad61682ff6b73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/89ba4c549bd628bc6c213302dd02250f904b0343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/89ba4c549bd628bc6c213302dd02250f904b0343"}]},"branch":"refs/heads/master"},"88a9d093b1001ca306b51f087a5c8fc882fc7ec2":{"kind":"REWORK","_number":4,"created":"2022-03-15 07:37:08.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/4"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/88a9d093b1001ca306b51f087a5c8fc882fc7ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/88a9d093b1001ca306b51f087a5c8fc882fc7ec2"}]},"branch":"refs/heads/master"},"7b7025013e039a8295365db1b4aeb31739bcb557":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-03-15 10:22:01.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/5"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 10:19:00.000000000","tz":330},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nDepends-On: Idae3cd84e8e3ac152d9c34521520dd01366001e0\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7b7025013e039a8295365db1b4aeb31739bcb557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7b7025013e039a8295365db1b4aeb31739bcb557"}]},"branch":"refs/heads/master"},"41639ea043be42d38bf2ea69b81a12be6707a9c8":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-03-15 13:31:51.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/6"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","date":"2022-03-15 13:31:51.000000000","tz":0},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nDepends-On: Idae3cd84e8e3ac152d9c34521520dd01366001e0\nDepends-On: I54504303490918a1d50a3f3cadc0d75ce0c84f33\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/41639ea043be42d38bf2ea69b81a12be6707a9c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/41639ea043be42d38bf2ea69b81a12be6707a9c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f21c4913feae4445df51ef371c0a670fa08eb6a9":{"kind":"REWORK","_number":7,"created":"2022-03-16 12:13:46.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/7"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-16 12:13:36.000000000","tz":330},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nDepends-On: Idae3cd84e8e3ac152d9c34521520dd01366001e0\nDepends-On: I54504303490918a1d50a3f3cadc0d75ce0c84f33\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f21c4913feae4445df51ef371c0a670fa08eb6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f21c4913feae4445df51ef371c0a670fa08eb6a9"}]},"branch":"refs/heads/master"},"8600dcdf3552253128859167a240160e7d77aa9f":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-03-16 13:24:49.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/8"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","date":"2022-03-16 13:24:49.000000000","tz":0},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8600dcdf3552253128859167a240160e7d77aa9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8600dcdf3552253128859167a240160e7d77aa9f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"64a03c280ff1c3cda0546ff19f7ceebcc5ed9b62":{"kind":"REWORK","_number":9,"created":"2022-03-16 17:58:13.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/9"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-16 17:57:35.000000000","tz":330},"subject":"Update setup_clients() api","message":"Update setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64a03c280ff1c3cda0546ff19f7ceebcc5ed9b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/64a03c280ff1c3cda0546ff19f7ceebcc5ed9b62"}]},"branch":"refs/heads/master"},"3cf94d721a03e6701fed790b47157b12517dfba6":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-03-16 18:15:17.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/10"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","date":"2022-03-16 18:15:17.000000000","tz":0},"subject":"Remove setup_clients() api","message":"Remove setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3cf94d721a03e6701fed790b47157b12517dfba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3cf94d721a03e6701fed790b47157b12517dfba6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"315fe69cc24ea9705d8ab39cca6e417dd958978a":{"kind":"REWORK","_number":11,"created":"2022-04-12 10:40:21.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/07/829207/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/07/829207/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/07/829207/11"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-15 07:34:15.000000000","tz":330},"committer":{"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","date":"2022-04-12 10:40:21.000000000","tz":0},"subject":"Remove setup_clients() api","message":"Remove setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: I38616a6be43619abf7e5d51e4884722dec4369a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/315fe69cc24ea9705d8ab39cca6e417dd958978a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/315fe69cc24ea9705d8ab39cca6e417dd958978a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
