)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~834455","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Iae8eac5dffe3c5d15e38a6acd69f2bad95f3f238","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"fix/isolated-networks","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2022-06-01 19:16:21.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-03-28 16:49:20.000000000","reason":"Douglas Mendizábal replied on the change"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-06-01 18:47:29.000000000","reason":"Ade Lee replied on the change"}},"hashtags":[],"change_id":"Iae8eac5dffe3c5d15e38a6acd69f2bad95f3f238","subject":"Enable isolated networks in tempest.conf","status":"MERGED","created":"2022-03-21 11:29:42.000000000","updated":"2022-06-01 19:16:21.000000000","submitted":"2022-06-01 19:16:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"834455-fix/isolated-networks","meta_rev_id":"2efcb96b25021f18e30c085a2c8bab0546abe5cc","_number":834455,"virtual_id_number":834455,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-01 19:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-03-28 16:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-04-26 08:40:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2022-06-01 18:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2022-06-01 18:47:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 12:30:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 10:16:35.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2022-03-24 10:16:35.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-03-24 10:16:35.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2022-03-28 16:49:20.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"CC"},{"updated":"2022-03-28 16:49:20.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2022-04-26 08:40:39.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"53cc77b23390b0c4dbb38813ab3d4099a794b98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-21 11:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"087bc77ceca551d6309af9efa818b02542d9f812","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-21 11:29:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"edd9c665dfb21b4d6b07101d8ada8ad2a3f9e300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:30:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4119dce0bf4d838859860544f5d889 : SUCCESS in 4m 26s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d96c358dc34a4dadba07ecf437610914 : SUCCESS in 40m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/27f29a5342a24cd8a486fdd2b4b2b31f : SUCCESS in 44m 00s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/9104f3d98e764b4d978a0872dce07ae6 : SUCCESS in 45m 57s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/fba1cc0076f246669e019cbb369aab80 : SUCCESS in 50m 24s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/5c7d856060414a21bee5b0490fbd65d7 : SUCCESS in 44m 28s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/8bd88a9872774e6cbfb9c771bd5aa404 : SUCCESS in 42m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/8337b5d47d1c43a5a85d4cbe2843bbe9 : SUCCESS in 53m 48s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/cd16aa4c98c84900ae69aec70c6dd879 : SUCCESS in 43m 22s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e24fd21dff114eefb8e9907ef0ec8128 : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"302e46595c8ca8788351dbdd89c1a7aaeb30fac6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-24 09:31:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45793198c091e11789fa63c97576bc0efc70d7b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 10:13:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160b53342c204a938562ec86d3730788 : SUCCESS in 4m 33s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1eba890aac15468da7bc90d17ad4565a : SUCCESS in 39m 05s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/2eda38c1f2184266bdd1400bba5721d8 : SUCCESS in 38m 18s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/db401774039d480392bf82279b7baa9d : SUCCESS in 35m 23s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/4431bc9ada374c7e8f07fb9dc011d423 : SUCCESS in 40m 00s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/ec1c1767c8714fe1bf507ceebddc5c2f : SUCCESS in 37m 26s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/145398cd155443db90d1425f3bfe857e : SUCCESS in 40m 22s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ca3ca8a0fff24a40abac55fe51f4107c : SUCCESS in 36m 52s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/301dd2a340b643a68800c627f5013731 : SUCCESS in 38m 17s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8c95a97986924ce3bc34aae4df2d99fd : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0f855d9c6e70e743fe6eae14fe67588fe7d7b2c3","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-24 10:16:35.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"11e685293b1cf72e372b7d5e2dc39f1262060fb1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-03-24 14:10:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"32891794ffcf73644dfef30a7401faed444b24de","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-03-24 14:25:14.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c83bf573aa228de5ee4c0e2a3dccbfd67ed0be75","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-28 13:27:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e24bdbe950fca349f2bb275bed4fb0a786d540f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-28 13:28:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1d70048f359ed3cf24c9f20a8c46a636422ee2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 14:28:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7548ef09c9934d78981175a57646324d : SUCCESS in 5m 04s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/86ebf350de5043efa0972d6faeaee917 : SUCCESS in 44m 22s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/2b1f01d3b4f84cf9a7848ced3fc468ed : SUCCESS in 51m 26s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/b7e37379079c4151b28e4c49c74dc8cb : SUCCESS in 41m 10s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/478bbe23f41240069abe99ce720b98aa : SUCCESS in 48m 36s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/ef64e31121534c5a9f9d74b2f6c186f5 : SUCCESS in 46m 56s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/7902b364a87144b4868dff096768cbb4 : SUCCESS in 45m 43s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/19c5de86649148e883569ee9fee92194 : SUCCESS in 45m 39s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/95b74c85501a4b7b84820f11d4b05ecc : SUCCESS in 47m 53s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/9ceda687abc3458e97fd73fb87daa819 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ae8e0ea0bdf24bf507f2008a2723a7c64ad6289e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-03-28 16:49:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1e41e59c1105f12ae2fa85e72644b7603da314ce","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-26 08:40:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d2ed7eede3d9254bf54124bc6f31df6365009d26","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-01 18:47:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e5ebcb8a563727e2d2e115c4dcc0ee52a224c768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 18:48:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ce609a89e1b490ce72d212f26aaa2d056bd2bf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 19:16:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dd7aeeb7894ddd91082de1f7154388 : SUCCESS in 2m 36s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a3f02636c4c844be8266d8f9ddfdd341 : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2efcb96b25021f18e30c085a2c8bab0546abe5cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 19:16:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11c8cfae0ce85b874ab055e685accba85f460fda","revisions":{"2d0c0f240bd230b7070798a2ed271e959910cb82":{"kind":"REWORK","_number":1,"created":"2022-03-21 11:29:42.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/55/834455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/55/834455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/1"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-03-21 11:28:49.000000000","tz":60},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-03-21 11:29:40.000000000","tz":60},"subject":"DNM: Isolated networks test","message":"DNM: Isolated networks test\n\nChange-Id: Iae8eac5dffe3c5d15e38a6acd69f2bad95f3f238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2d0c0f240bd230b7070798a2ed271e959910cb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2d0c0f240bd230b7070798a2ed271e959910cb82"}]},"branch":"refs/heads/master"},"56266ec49c06ad0b7d85995e30a60420f3089551":{"kind":"REWORK","_number":2,"created":"2022-03-24 09:31:24.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/55/834455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/55/834455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/2"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-03-21 11:28:49.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2022-03-24 09:31:12.000000000","tz":0},"subject":"Enable isolated networks in tempest.conf","message":"Enable isolated networks in tempest.conf\n\nThere was an issue with s-rbac and tempest when isolated networks\nwere enabled in tempest.conf.\n\nThis is no longer an issue as test that were failing with\ncreate_isolated_networks\u003dtrue and s-rbac enabled are now\npassing successfully.\n\nChange-Id: Iae8eac5dffe3c5d15e38a6acd69f2bad95f3f238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/56266ec49c06ad0b7d85995e30a60420f3089551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/56266ec49c06ad0b7d85995e30a60420f3089551"}]},"branch":"refs/heads/master"},"11c8cfae0ce85b874ab055e685accba85f460fda":{"kind":"REWORK","_number":3,"created":"2022-03-28 13:28:27.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/55/834455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/55/834455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/55/834455/3"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-03-21 11:28:49.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2022-03-28 13:22:56.000000000","tz":0},"subject":"Enable isolated networks in tempest.conf","message":"Enable isolated networks in tempest.conf\n\nThere was an issue with s-rbac and tempest when isolated networks\nwere enabled in tempest.conf.\n\nThis is no longer an issue as test that were failing with\ncreate_isolated_networks\u003dtrue and s-rbac enabled are now\npassing successfully.\n\nChange-Id: Iae8eac5dffe3c5d15e38a6acd69f2bad95f3f238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/11c8cfae0ce85b874ab055e685accba85f460fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/11c8cfae0ce85b874ab055e685accba85f460fda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
