)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~841014","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I9d36100266e63aedca256a37fc364e089157dab1","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-12-02 16:40:57.000000000","reason":"\u003cGERRIT_ACCOUNT_34293\u003e replied on the change","reason_account":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"}}},"removed_from_attention_set":{"34293":{"account":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"},"last_update":"2022-12-02 16:40:57.000000000","reason":"\u003cGERRIT_ACCOUNT_34293\u003e replied on the change","reason_account":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"}}},"hashtags":[],"change_id":"I9d36100266e63aedca256a37fc364e089157dab1","subject":"Add endpoint_type in configuration","status":"NEW","created":"2022-05-07 14:31:34.000000000","updated":"2022-12-02 16:40:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dec7fda3a0563f7ece4d97e7309c9ce853c9ed4e","_number":841014,"virtual_id_number":841014,"owner":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-07 15:13:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":-1,"date":"2022-10-19 20:12:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-07 15:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 20:12:14.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"33d25d890399a1be81a70f147ed73b9a04ebe468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"},"date":"2022-05-07 14:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5792121378056783290a7ba1a9f03cc608e9beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 15:13:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be825e6cddf4b7492f75cda1384f7b3 : SUCCESS in 2m 56s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/646373ec8830465c907b3a88e5adbaf5 : SUCCESS in 25m 51s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/394c203f78fb4d32968f5916a2bf7920 : SUCCESS in 32m 56s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/5832960949b24bd781def8b50638407a : SUCCESS in 31m 50s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/ac33462ac6d24086b41a750ab6eb1bc3 : SUCCESS in 40m 07s\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/feed40f442524a42a60a35de07d02bf1 : SUCCESS in 36m 48s\n- barbican-tempest-plugin-simple-crypto-ussuri https://zuul.opendev.org/t/openstack/build/886c15e16cba4949bcb21eeb3429fb1f : SUCCESS in 36m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/8c7cece9b5124cf3b09f2224edc0d21b : SUCCESS in 35m 47s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/40dd34e7acee48019deada4f6154798a : SUCCESS in 36m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f642fe8eb1694520ae7ed98369c59bc5 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"acafc7b96f52511f621af936c905e0192c940b32","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-10-19 20:12:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa3ce0a241f5f3c7c2d8d9afae50f14b86f27095","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-10-27 14:26:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dec7fda3a0563f7ece4d97e7309c9ce853c9ed4e","author":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"},"date":"2022-12-02 16:40:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"091e1e86e28715d099a00c00d1fed4a59bb1a963","revisions":{"091e1e86e28715d099a00c00d1fed4a59bb1a963":{"kind":"REWORK","_number":1,"created":"2022-05-07 14:31:34.000000000","uploader":{"_account_id":34293,"name":"Jose Bautista Barato","email":"jose.bautista.barato@gmail.com","username":"josebb"},"ref":"refs/changes/14/841014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/841014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/841014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/841014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/841014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/841014/1"}}},"commit":{"parents":[{"commit":"9fe096e1422c7c884b94166763b4857b5abcea30","subject":"Merge \"Add secure-rbac test for Consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9fe096e1422c7c884b94166763b4857b5abcea30"}]}],"author":{"name":"josebb","email":"jose.bautista.barato@gmail.com","date":"2022-05-07 14:30:37.000000000","tz":180},"committer":{"name":"josebb","email":"jose.bautista.barato@gmail.com","date":"2022-05-07 14:31:30.000000000","tz":180},"subject":"Add endpoint_type in configuration","message":"Add endpoint_type in configuration\n\nChange-Id: I9d36100266e63aedca256a37fc364e089157dab1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/091e1e86e28715d099a00c00d1fed4a59bb1a963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/091e1e86e28715d099a00c00d1fed4a59bb1a963"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
