)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~854250","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I16ed904eeb27ab7110a7e4e56ef7ea89c8c3c2ab","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"story/2010235","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-24 13:12:40.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-08-24 13:12:40.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-08-24 12:25:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9914\u003e replied on the change","reason_account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}},"hashtags":[],"change_id":"I16ed904eeb27ab7110a7e4e56ef7ea89c8c3c2ab","subject":"Add \"member\" role to tempest tests","status":"MERGED","created":"2022-08-23 18:20:47.000000000","updated":"2022-08-24 13:12:40.000000000","submitted":"2022-08-24 13:12:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854250-story/2010235","meta_rev_id":"16ee064b6d013dee72c6a6f2a37456f65298e672","_number":854250,"virtual_id_number":854250,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 13:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-08-24 12:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-08-24 12:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 19:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 00:38:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-08-24 00:38:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"e131112df4d94c535f7e1761b9bfeb7024b569c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-23 18:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309c5e11bef69002c644bcca221347b73e897e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 19:16:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a5cce64e884789b3b6d4cd1ee5ba48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/119ffd8084d446a38e489ea1b35818bc : SUCCESS in 3m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/667bdc179b2048dcbf113b9f66dcbfd6 : SUCCESS in 39m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9ca0b50792cb4eec9729b1b697ee82d8 : SUCCESS in 54m 18s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/dd048afa7d104f86aabc6cb3d6e74e4b : SUCCESS in 45m 51s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/3a28285a1a464d509de3fa3993008f09 : SUCCESS in 38m 10s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/dfdbcd215ae54928b85ad0a2e95ad7e5 : SUCCESS in 43m 39s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/1e12792f3374456c9751dafe03e9885e : SUCCESS in 41m 51s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/13535ba807d047cfadf6dff88cea8f8f : SUCCESS in 45m 15s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/d5213efd2d0c42acb7a5d4529b3ba18c : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4eb7d364d4750c82d9b2d9b94f4beb1b70b87f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-23 19:57:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50420c3cbc98d24830ffa73a0dce950d76773007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:40:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a1300051fe4854818228179f813d7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33bcf116ec024000b842caca1dc10ff0 : SUCCESS in 3m 04s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/35acb0bc9f474c088f12cbaaf94a62fb : SUCCESS in 36m 45s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/57214b33ff314e7497ce5d664aabab44 : SUCCESS in 37m 52s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/8804297613ad4dbf854a37129f985619 : SUCCESS in 42m 04s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/c8a136ec98294547aa41767377795aee : SUCCESS in 36m 29s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/e6ac8e3e28104a668d03ab9db26945fb : SUCCESS in 38m 22s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/3e881289b6644c28af335ba6e6107e3d : SUCCESS in 38m 55s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/3791af3121464d6ab62e60ca3456e4e0 : SUCCESS in 37m 15s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/4032120a61104322ae7390280fa08569 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":2},{"id":"24428f6e700ef706ea39c05a7823d39eaad50605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-23 21:29:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"085eaad679db096e71e97fabc84c4ad14abf511b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 22:30:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ad12b6b5ae241bda186efd5259db37b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115180de0cbb46e1aec3f4f4998dcb2f : SUCCESS in 3m 18s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8095bc1b6e8e42f0a0776848bcb232eb : SUCCESS in 57m 59s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9b320293ee7a4e408d399857bb49bcca : SUCCESS in 59m 30s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/2734d6c938f34e41862c1356b3b20498 : SUCCESS in 55m 27s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/911a6d5937aa456db1cc132a5ba09048 : SUCCESS in 52m 52s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/05caef9e67a04452881948e2c35bc4db : SUCCESS in 35m 30s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/632d4231402f4f65afc940691e8f9f17 : SUCCESS in 58m 11s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/0faea657c7cc4643a1a0c7365993f105 : SUCCESS in 45m 24s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/bebd68bacd004fc2ac07490dff85dee7 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a19c34f42b629cde4f762e848d0be9a868e5b0f2","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-24 12:25:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2e5b67249c26462c74b3f95470e9b5d9d0874df7","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-24 12:26:03.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"91f675070aad7c0ae3ce4f056bfe5bc179592f67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 12:26:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9743862112b5af0215871a8d8f56a4f27b7ba630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 13:12:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240d9acfa96c4be8891dc34dc2e0a349\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c0b2d0164240a5a7af006246f8597e : SUCCESS in 3m 07s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6781dde7ae01449d8f48add2860cb727 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":3},{"id":"16ee064b6d013dee72c6a6f2a37456f65298e672","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 13:12:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0aa2f34d0bac13d1706145a510d4474918caf6d","revisions":{"724dfea918387a56ca352182fbc954c7147774da":{"kind":"REWORK","_number":1,"created":"2022-08-23 18:20:47.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/50/854250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/50/854250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/1"}}},"commit":{"parents":[{"commit":"45dd1313926858b487b840701ed5e449d44e37b7","subject":"Add RBAC tests for secret consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/45dd1313926858b487b840701ed5e449d44e37b7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 18:18:27.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 18:18:27.000000000","tz":-240},"subject":"Add \"member\" role to tempest tests","message":"Add \"member\" role to tempest tests\n\nThe \"member\" role is needed by the scenario tests to create secrets when\nSecure RBAC is enabled i.e. enable_scope\u003dTrue\n\nChange-Id: I16ed904eeb27ab7110a7e4e56ef7ea89c8c3c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/724dfea918387a56ca352182fbc954c7147774da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/724dfea918387a56ca352182fbc954c7147774da"}]},"branch":"refs/heads/master"},"8e98e739307d73e7bfc71190b8c335aac0397b8f":{"kind":"REWORK","_number":2,"created":"2022-08-23 19:57:47.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/50/854250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/50/854250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/2"}}},"commit":{"parents":[{"commit":"45dd1313926858b487b840701ed5e449d44e37b7","subject":"Add RBAC tests for secret consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/45dd1313926858b487b840701ed5e449d44e37b7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 18:18:27.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 19:57:29.000000000","tz":-240},"subject":"Add \"member\" role to tempest tests","message":"Add \"member\" role to tempest tests\n\nThe \"member\" role is needed by the scenario tests to create secrets when\nSecure RBAC is enabled i.e. enable_scope\u003dTrue\n\nChange-Id: I16ed904eeb27ab7110a7e4e56ef7ea89c8c3c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8e98e739307d73e7bfc71190b8c335aac0397b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8e98e739307d73e7bfc71190b8c335aac0397b8f"}]},"branch":"refs/heads/master"},"d0aa2f34d0bac13d1706145a510d4474918caf6d":{"kind":"REWORK","_number":3,"created":"2022-08-23 21:29:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/50/854250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/50/854250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/50/854250/3"}}},"commit":{"parents":[{"commit":"45dd1313926858b487b840701ed5e449d44e37b7","subject":"Add RBAC tests for secret consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/45dd1313926858b487b840701ed5e449d44e37b7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 18:18:27.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-23 21:29:19.000000000","tz":-240},"subject":"Add \"member\" role to tempest tests","message":"Add \"member\" role to tempest tests\n\nThe \"member\" role is needed by the scenario tests to create secrets when\nSecure RBAC is enabled i.e. enable_scope\u003dTrue\n\nChange-Id: I16ed904eeb27ab7110a7e4e56ef7ea89c8c3c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d0aa2f34d0bac13d1706145a510d4474918caf6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d0aa2f34d0bac13d1706145a510d4474918caf6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[]}
