)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~854614","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Iae1dfea83a6584a95fb73488ad0ba988ded371ca","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"story/2010235","attention_set":{},"removed_from_attention_set":{"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-08-25 22:02:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae1dfea83a6584a95fb73488ad0ba988ded371ca","subject":"Add versioned jobs","status":"MERGED","created":"2022-08-25 12:00:19.000000000","updated":"2022-08-25 22:02:15.000000000","submitted":"2022-08-25 22:02:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854614-story/2010235","meta_rev_id":"10c23c55bcacd8ce48c513caddaa581bddc452f6","_number":854614,"virtual_id_number":854614,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 22:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-08-25 21:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-08-25 21:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 13:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-25 21:05:37.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"9557f3adeb1090e7bb2c150faba59778e72e01e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 12:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd2937ab0c45766b6b90cd3277e3a9401248522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 13:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5321838679645b7a541e0bf0d60644b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c37b1a9ea8405abed3cbfc6e550e32 : SUCCESS in 2m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3469073f49d4417abeb5c758ca29f1f1 : SUCCESS in 47m 42s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/5a7cb98d3a104f9e8ea75ae2459babd8 : SUCCESS in 58m 47s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/96c5e936ec5744a8bf296cb2f1c236f4 : SUCCESS in 49m 15s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/1dfaf54acc8d4e6e9fec5a7bd33f040a : SUCCESS in 48m 57s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/3b17e02684f741f2bd3a1ab363d2e08a : SUCCESS in 50m 45s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c5d457fa57dd43688fb7c354fafa687f : SUCCESS in 47m 06s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f91c9b3c2b154e3991f8d437b619d9af : SUCCESS in 50m 06s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/672962e2d7f34742a6e2c5ccb0d71644 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4160123a34a56f001aae8249159e76e7b8e9033a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-25 21:05:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4701b10350051fc511070a27513240af005c5bba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 21:06:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f88e3392d68eedbe815744d8cbd33150cf48865b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-25 21:07:11.000000000","message":"Topic fix-srbac-consumer-tests removed","accounts_in_message":[],"_revision_number":1},{"id":"902300e2748abba560f0a2a4e6016b6ea17e090c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-25 21:07:31.000000000","message":"Topic set to story/2010235","accounts_in_message":[],"_revision_number":1},{"id":"406a7513c3d27ebe7844d1a6c428d8c41d8e72a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:02:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad790631b3494f28b9811d5ca71575f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6f2a3c3dbb4923a739f662882d309f : SUCCESS in 3m 04s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e04782d5805b419589272613d83678a6 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":1},{"id":"10c23c55bcacd8ce48c513caddaa581bddc452f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:02:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01da5615dbe869685858f64e52fe40c870141116","revisions":{"01da5615dbe869685858f64e52fe40c870141116":{"kind":"REWORK","_number":1,"created":"2022-08-25 12:00:19.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/14/854614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/854614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/854614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/854614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/854614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/854614/1"}}},"commit":{"parents":[{"commit":"d0aa2f34d0bac13d1706145a510d4474918caf6d","subject":"Add \"member\" role to tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d0aa2f34d0bac13d1706145a510d4474918caf6d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 11:58:44.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 11:58:44.000000000","tz":120},"subject":"Add versioned jobs","message":"Add versioned jobs\n\nWith 1.1 coming in zed, we need to version the old jobs to prevent\n1.1 tests (secret consumers) being run and failing on old branches.\n\nChange-Id: Iae1dfea83a6584a95fb73488ad0ba988ded371ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01da5615dbe869685858f64e52fe40c870141116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01da5615dbe869685858f64e52fe40c870141116"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[]}
