)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~882703","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I91874ec6001d82104630b4f33883b5af9d2ac09f","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"barbican-launchpad","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-05-19 16:57:53.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-05-19 17:18:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91874ec6001d82104630b4f33883b5af9d2ac09f","subject":"Migrate back to Launchpad","status":"MERGED","created":"2023-05-09 11:55:54.000000000","updated":"2023-05-19 17:18:50.000000000","submitted":"2023-05-19 17:18:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882703-barbican-launchpad","meta_rev_id":"3fced8a1a5591b3a0e47a7eb229ca56e36e600d2","_number":882703,"virtual_id_number":882703,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 17:18:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2023-05-19 16:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2023-05-19 16:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 12:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 12:23:04.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"efc05de408fde1074d8f2a43ca0e1909aad1ece6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-05-09 11:55:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce34000da721558b6aeeca8ddaa2eaf5b03cab01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0511df74bf94aa3bb2dc0aa0ef0f5ed\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e23e90a5fba4d0ebee438a2bf4848c0 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff44e327db63428ca0d1e6d980c7c0f8 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6bab75d8615e27134c13009eeda52e28fd6b972e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-05-09 13:37:57.000000000","message":"Topic set to barbican-launchpad","accounts_in_message":[],"_revision_number":1},{"id":"70779c646870a3bb8da1cb381b37ef657cc2bdd9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-05-19 16:57:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"280ed1269fccbc3a960ae14b049543dd63205cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 16:58:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1ca1f038edec51fe5cec4e6652133c102b09e42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 17:18:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f4ab9e74e95431fa21ef67a6516dd87\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c15335c79f4e4acbb28a0dc1ad2e9ed9 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094581c8098043299deec040c6253ad3 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3fced8a1a5591b3a0e47a7eb229ca56e36e600d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 17:18:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42300cd1e56459fcbab6d98f74c9d5dd0442fe5c","revisions":{"42300cd1e56459fcbab6d98f74c9d5dd0442fe5c":{"kind":"REWORK","_number":1,"created":"2023-05-09 11:55:54.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/03/882703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/03/882703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/03/882703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/03/882703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/03/882703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/03/882703/1"}}},"commit":{"parents":[{"commit":"30e50ef9c85bf008c7c3e5aeb6b009629a2013c9","subject":"Merge \"Introduce a new test for \"cve_2022_3100\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/30e50ef9c85bf008c7c3e5aeb6b009629a2013c9"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-05-09 11:55:34.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-05-09 11:55:52.000000000","tz":120},"subject":"Migrate back to Launchpad","message":"Migrate back to Launchpad\n\nChange-Id: I91874ec6001d82104630b4f33883b5af9d2ac09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/42300cd1e56459fcbab6d98f74c9d5dd0442fe5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/42300cd1e56459fcbab6d98f74c9d5dd0442fe5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
