)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~882911","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-02-19 20:50:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-02-20 18:12:57.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-20 18:12:57.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-02-20 17:09:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46","subject":"Fix image_signature_verification group not register properly","status":"MERGED","created":"2023-05-11 03:08:40.000000000","updated":"2024-02-20 18:12:57.000000000","submitted":"2024-02-20 18:12:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882911","meta_rev_id":"d0d3ced9700865d2e70e716b53b548d2394d2c34","_number":882911,"virtual_id_number":882911,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-20 18:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-02-19 20:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-02-20 17:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-02-20 17:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 04:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-29 14:54:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-01-29 14:54:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-01-29 14:54:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"9a09c6ec686cfe2056fc4347d8dc9f2f18a6c31f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-11 03:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b9ef499ef877074797dd175bef367897e56b22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 04:10:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d046c78b9a34a308f68d2bc096b0103\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bff4324c9a94e88b1091d6b5ff43695 : SUCCESS in 2m 27s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f1983a41556d4a1e98af34510686a2a4 : SUCCESS in 40m 26s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/d7828370cd384af680cef8db3b427616 : SUCCESS in 55m 26s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/a9e44f04b8114f21afd6b2a9a0c41438 : SUCCESS in 46m 10s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/678f6e0b858b4a539101e39f31ee89f5 : SUCCESS in 33m 05s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/6e10a2e9bf394e2593343dc59a29ed56 : SUCCESS in 31m 59s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/926de1673fad4dc5b491baa091ce8416 : SUCCESS in 34m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/0685ff46f8124606a520091c209d97f1 : SUCCESS in 53m 04s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/baddc22d1d8d456882ca64d7ee387b0a : SUCCESS in 40m 11s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/7e5412187f7442bb9757ff6c70a5e7f6 : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dce56b7d2804b15d8745efad6d6b999a746e2846","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-01-31 05:16:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5dc814863b806cfda77168c77c2f6280002b0e98","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-02-09 15:23:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6fe0b2366298a8c6e86e4b76e432f912bf4b8e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 15:52:27.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/barbican-tempest-plugin for 882911,1","accounts_in_message":[],"_revision_number":1},{"id":"5dc76e92856cad30301809aa3e7c8a858ff880eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 16:14:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b85eea7cb4d5b908953788c79c02c9b7e0f2edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 17:01:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16b6d0a5c9d74d6c9e03f00a9bfae4ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa7bebcfa624d30a68b0bd231a7f1c7 : SUCCESS in 3m 52s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/14a46ab6646c47f5903a2a84e3f92a01 : FAILURE in 31m 10s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f0f1f042af4643189778d14cad3f4f06 : FAILURE in 37m 28s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/76e5d4735e0b4a02a97454967951f715 : FAILURE in 26m 19s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/cfff1d2bee7f48e4baef310d02716864 : FAILURE in 40m 34s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/de07e00cc4c743f8bc874f9c2eae014f : FAILURE in 29m 31s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/44c5dbb56bef415f84e32faf988299b6 : FAILURE in 22m 11s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/a7e7313f943b45ba82b8ac04b556ff99 : SUCCESS in 35m 52s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/32702256129f4df3b2cb1a1f82d4a8d5 : FAILURE in 27m 22s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/2caee90776de4553a40e82035627ca10 : FAILURE in 47m 03s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e22a92e5637a4d8a9e2c3477438604bd : FAILURE in 37m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6f4397042641614bd1aece9aa68bd34ee2ba0c43","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:41:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4ddc65e3410192893a0171c7b2e3832bfc00609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:42:35.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab94bc9a41f24bd649b42c0cb5e041b3e2d1e7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:44:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3bff1a5bb625dd7fcbcda575e5efbe401731645d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:42:18.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6f67195a7df41a5b166c641bcd8ce27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37dcceaa8c9e417ca49588105c418839 : FAILURE in 2m 36s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6bbc8d550c9f4e25ba9875086fde2029 : FAILURE in 27m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/00de1882b17f4334a8ec9d0c48755352 : FAILURE in 27m 53s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/f44aa76f2e64416e8fbe1a93ebfa4663 : FAILURE in 24m 27s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/6518fd8b51ce4b4ea39a7aed862441ac : FAILURE in 41m 22s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/101ed4f9c18c4fcfb2e28d2fd18074da : FAILURE in 14m 53s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/ad3483b46c4f4fb292f619278bdb71f2 : FAILURE in 41m 10s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/92648062be3140ae964e11a4b80e81c4 : SUCCESS in 49m 44s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/98ab8dfc3fdf48aeaa8245b45e5720ca : FAILURE in 37m 41s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/36ba9308b7704ddf8d1ce4b61e426e38 : FAILURE in 38m 17s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e1973f57fb90425cac326e2e1b3c9970 : FAILURE in 27m 38s","accounts_in_message":[],"_revision_number":4},{"id":"fe917797f6ada0cc9238e71884a337e10662cb9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 16:44:15.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"facbc9085b181475fee8d10c6c649b2a775f8476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 16:46:22.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"388c6c76f29c9060e0a668b7625b2077a870d386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:37:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28475a333fa84b1b8da6c08f6f7d6852\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18af590e4e8b4afda0ac9114c73f5e70 : SUCCESS in 2m 53s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b07a3714d98e476491bfdf39628bab9c : SUCCESS in 41m 44s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/5b5aeffece9d4fbfa3a829a0f3bbe183 : SUCCESS in 43m 35s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/6564bb526b42467a804fe1f6d07dd7f0 : SUCCESS in 28m 40s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/7aa3d48d373d44f9b9b839bafeb0907a : SUCCESS in 38m 52s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/f441025484a144b4b35a6b1440723833 : FAILURE in 20m 14s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/8ac028384de74c2caf29cfaf2d2e7220 : SUCCESS in 31m 15s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/39c206e2fa744e6d96370484330f50c9 : SUCCESS in 41m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/bd3f73c557634ac8a74fd89fc83a1326 : SUCCESS in 45m 01s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/858421d36380455189a86a34d036ccd1 : SUCCESS in 41m 56s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/ff88cb4080344a23af388f930c3d79e2 : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":6},{"id":"ba38750eb45d9c788217775b1624831476664c11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-19 19:45:41.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":7},{"id":"6ad3b57f256f946912c007fcc5df8ff4ad9eef2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 20:36:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d325158b5881463d89380d495750663f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8fb8d5717c47b08463c21ff6d747e8 : SUCCESS in 3m 50s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8707d93f38074dd6bc718b7750841866 : SUCCESS in 39m 01s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/aefb0de5065c4fa59e7792b4ae99f172 : SUCCESS in 43m 14s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/7086b667857a4abaabb71397f7ee6172 : SUCCESS in 41m 26s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/4a2113b0333e4c3491ee733915b11a86 : SUCCESS in 43m 13s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/d2bc86dc80c843a3816f87003d8b8a6c : SUCCESS in 43m 17s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/260e8ef7b9c24584b436be3578e22e7a : SUCCESS in 44m 25s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/20b99800448144729d6d23d0277701ab : SUCCESS in 44m 16s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/452e6081821b4d3a90018fb6a1812765 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":7},{"id":"07247221e7d8aedf2dcaed56d9161d0ef62cb65a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-19 20:50:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2d5ffcc75b42d7cc0a201c67203deb0dfc04042e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-02-20 17:09:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7be42ca37f556dd2fed26cc7f3e1b55901552b75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 17:09:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6ceccca3543cefb996ab9f1172413eeb0a47946a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 18:12:57.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc3500a355c64c368015c568bfe2aaf4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff32e951e0cf4417b4af9123371e03a5 : SUCCESS in 2m 31s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a4bd36f87b064bf8ab1894c45ef9d1fc : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":7},{"id":"d0d3ced9700865d2e70e716b53b548d2394d2c34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 18:12:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8f696b3219d0c30bbcd8ed89570a37de42a101b5","revisions":{"47cea389e4e50c4903f2572a641473d80aef9b40":{"kind":"REWORK","_number":1,"created":"2023-05-11 03:08:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/1"}}},"commit":{"parents":[{"commit":"30e50ef9c85bf008c7c3e5aeb6b009629a2013c9","subject":"Merge \"Introduce a new test for \"cve_2022_3100\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/30e50ef9c85bf008c7c3e5aeb6b009629a2013c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:39.000000000","tz":540},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/47cea389e4e50c4903f2572a641473d80aef9b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/47cea389e4e50c4903f2572a641473d80aef9b40"}]},"branch":"refs/heads/master"},"b627178f1db212fc94e3acc31e49a99aa9a54baf":{"kind":"REWORK","_number":2,"created":"2024-02-09 16:14:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/2"}}},"commit":{"parents":[{"commit":"f2b9c024cb736551668bc871201d71542896b508","subject":"Merge \"Replace deprecated tenant_id property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2b9c024cb736551668bc871201d71542896b508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 16:13:46.000000000","tz":540},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b627178f1db212fc94e3acc31e49a99aa9a54baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b627178f1db212fc94e3acc31e49a99aa9a54baf"}]},"branch":"refs/heads/master"},"ff946b305c92f7500fc4763914050441b94f831c":{"kind":"REWORK","_number":3,"created":"2024-02-16 15:42:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/3"}}},"commit":{"parents":[{"commit":"f2b9c024cb736551668bc871201d71542896b508","subject":"Merge \"Replace deprecated tenant_id property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2b9c024cb736551668bc871201d71542896b508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 15:42:34.000000000","tz":0},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ff946b305c92f7500fc4763914050441b94f831c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ff946b305c92f7500fc4763914050441b94f831c"}]},"branch":"refs/heads/master"},"784ebf541bfbc968ad6c9c86a4bd1c0509484e9e":{"kind":"REWORK","_number":4,"created":"2024-02-16 15:44:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/4"}}},"commit":{"parents":[{"commit":"f2b9c024cb736551668bc871201d71542896b508","subject":"Merge \"Replace deprecated tenant_id property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2b9c024cb736551668bc871201d71542896b508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 15:44:05.000000000","tz":540},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/784ebf541bfbc968ad6c9c86a4bd1c0509484e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/784ebf541bfbc968ad6c9c86a4bd1c0509484e9e"}]},"branch":"refs/heads/master"},"ef4f51d194a188cda9f87b6e1e6c27a8146db391":{"kind":"REWORK","_number":5,"created":"2024-02-16 16:44:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/5"}}},"commit":{"parents":[{"commit":"f2b9c024cb736551668bc871201d71542896b508","subject":"Merge \"Replace deprecated tenant_id property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2b9c024cb736551668bc871201d71542896b508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 16:44:15.000000000","tz":0},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ef4f51d194a188cda9f87b6e1e6c27a8146db391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ef4f51d194a188cda9f87b6e1e6c27a8146db391"}]},"branch":"refs/heads/master"},"691db68303ad511fe79ac27d676e54b5e7f65597":{"kind":"REWORK","_number":6,"created":"2024-02-16 16:46:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/882911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/6"}}},"commit":{"parents":[{"commit":"f2b9c024cb736551668bc871201d71542896b508","subject":"Merge \"Replace deprecated tenant_id property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f2b9c024cb736551668bc871201d71542896b508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 16:46:21.000000000","tz":0},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/691db68303ad511fe79ac27d676e54b5e7f65597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/691db68303ad511fe79ac27d676e54b5e7f65597"}]},"branch":"refs/heads/master"},"8f696b3219d0c30bbcd8ed89570a37de42a101b5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-19 19:45:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/11/882911/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/11/882911/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/11/882911/7"}}},"commit":{"parents":[{"commit":"c8b6d07de8eaee21a39bd8fd56087444c1b202c0","subject":"Update stable branch testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c8b6d07de8eaee21a39bd8fd56087444c1b202c0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-11 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-19 19:45:41.000000000","tz":0},"subject":"Fix image_signature_verification group not register properly","message":"Fix image_signature_verification group not register properly\n\nChange-Id: I32d4a5864c2cf9ec364167c6f7dada21a1a6bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8f696b3219d0c30bbcd8ed89570a37de42a101b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/8f696b3219d0c30bbcd8ed89570a37de42a101b5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
