)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~891314","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I780baefa1be94b999ef17df19798ed5bddada437","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{"34934":{"account":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"last_update":"2023-08-17 07:09:44.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I780baefa1be94b999ef17df19798ed5bddada437","subject":"Cleanup py27 support","status":"NEW","created":"2023-08-17 06:09:24.000000000","updated":"2023-08-17 07:09:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffc5978fd0905e417e41f775397ea4ae886f36a8","_number":891314,"virtual_id_number":891314,"owner":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-17 07:09:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 07:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71dbd24c8b975db2b476b49a54bed3b5f6d337d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-08-17 06:09:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffc5978fd0905e417e41f775397ea4ae886f36a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:09:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d0f7c6109b24483aae311d2b4cd7223\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f2f85eb68a840248e7968b1c118963b : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93b718974824431a6981a0531c5a2a9 : SUCCESS in 2m 31s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4b147b3e19db472eab1a587abf5ba4a9 : SUCCESS in 35m 59s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/55f91f9caa0b475199ab01300e6cb532 : FAILURE in 42m 31s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/4827e6c090f5442e9af878e1fb25e89f : SUCCESS in 42m 22s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/46a7489c82f340338b00993f47a31249 : SUCCESS in 39m 39s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/caeae4fdaf01473a85491e1ced9162c4 : SUCCESS in 37m 09s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/cd6656b92c7a498d98efbc66f5c65ece : SUCCESS in 41m 55s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/f6f541c570aa4e1a985d59f94c0c4207 : SUCCESS in 33m 24s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/95f83e00dd5c4fe587f62ed071013c3c : SUCCESS in 51m 57s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/0999d2af0a354522847880c223e0fa2d : SUCCESS in 37m 08s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/528b2b6b7e084beeb4debadbda4ed831 : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3555e656c039a7eee42d8010efd78a60125b4aeb","revisions":{"3555e656c039a7eee42d8010efd78a60125b4aeb":{"kind":"REWORK","_number":1,"created":"2023-08-17 06:09:24.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/14/891314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/14/891314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/891314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/891314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/891314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/14/891314/1"}}},"commit":{"parents":[{"commit":"bf83cf93c0063dde7078e8ff701e1712307fd1c0","subject":"Merge \"Add stable/2023.1 jobs on master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/bf83cf93c0063dde7078e8ff701e1712307fd1c0"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-08-17 06:09:07.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-08-17 06:09:07.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: I780baefa1be94b999ef17df19798ed5bddada437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3555e656c039a7eee42d8010efd78a60125b4aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3555e656c039a7eee42d8010efd78a60125b4aeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
