)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~899274","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"36388":{"account":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"last_update":"2024-01-31 05:55:04.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-01-31 05:29:34.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-01-22 15:59:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4","subject":"Barbican tempest missing IDs","status":"MERGED","created":"2023-10-25 14:56:20.000000000","updated":"2024-01-31 05:55:04.000000000","submitted":"2024-01-31 05:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"899274","meta_rev_id":"6f576d132ef9f129c8240e2957758372bdf01fd4","_number":899274,"virtual_id_number":899274,"owner":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 05:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"recommended":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"all":[{"value":1,"date":"2024-01-19 08:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":2,"date":"2024-01-31 05:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-01-24 08:14:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2024-01-22 15:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":1,"date":"2024-01-31 05:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 16:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 09:03:50.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2023-11-21 12:10:23.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-01-17 13:54:01.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2024-01-22 17:42:08.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-01-24 08:14:01.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"4bf17edd687a4418c4a14a25061fd9ed06b83146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-10-25 14:56:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc9017c0c7abac6126a47be1b3f2ebf588b698fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-10-25 15:27:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e8b8369d0bea385c2e052a341f7af38374fd8b8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-10-25 15:31:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3c06b85785497f3e7542e68007697d1e4005678d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-10-25 15:33:51.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"86e1c981be6f7ffbdf0a5552ef085ce68c4356a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 16:44:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/505adc2483464d62b5fc0b3f8f71d89c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49d5f89267c4fb481b9eb228a59da67 : SUCCESS in 3m 17s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0ce7455cc02043a08f9eaf7a861cd82d : SUCCESS in 33m 21s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/7e5fbc3c382c4e909e7a1ecb7d438e34 : FAILURE in 53m 30s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/245a6660066b4814a214e16c70e81e3f : SUCCESS in 32m 17s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/0b862fc4696345e998690030beba1c04 : SUCCESS in 1h 03m 30s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/b8494b56cc53451f950f3ad1666e18e0 : SUCCESS in 31m 08s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/a0691c43c16d4ca592cda833213f9170 : SUCCESS in 44m 32s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/e2337d97faca42928e8ede7f3865c03f : SUCCESS in 59m 57s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d8df7dc0c7de4637a7ef256ea7e16265 : SUCCESS in 32m 15s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/dd6ac0cf6b154b8bb3cd147d203de590 : SUCCESS in 29m 10s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/205b298581644b0c8752d1ea345293aa : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":4},{"id":"9373ad39da81e1a3c17b5175a344f86fc5fb7b74","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-10-27 09:03:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ffd9d062ebde0981f93666a032df50e30217b2c","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-11-01 15:37:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"75d951a913ee91164847fbd05b73429ca24fea99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-11-08 13:56:43.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7da51fd4ff9c6692ed5bc55e2d43f6c7e1d65d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-11-08 13:58:25.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"55b821b106427c5e1c59ff4071a0f7bed3fcb2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"date":"2023-11-08 13:58:48.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"3a56612062ace4320cb7868b9892cb8c5d92a871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:56:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6316255981c3414a9367ff39e16c15f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/111a4d09a09b46c8b7c09d17d3e74d21 : SUCCESS in 3m 05s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ac6d54b29c7f42e6ab6c51c954381ede : SUCCESS in 34m 51s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/51c7f5e66d3c42678f09c6af911c566c : FAILURE in 41m 56s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/4e158746eb8547be99223c0bb55b9ed0 : SUCCESS in 46m 30s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/cb9f917f5b9940cd8a4b70fb071e8a17 : SUCCESS in 33m 15s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/9bb5efca5e4e489bb6be473fd7b9c471 : SUCCESS in 32m 33s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/9061ac2d0e684d7189d0d1e857daf982 : SUCCESS in 37m 02s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/3b6236e038a14226aa745bce56ff8a4a : SUCCESS in 52m 15s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/f931fe409a154c43af71f099ed1668d8 : SUCCESS in 49m 09s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/73c96299e21c48d382eda3c8361acf4f : SUCCESS in 36m 40s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/7826aeedca4347f3add67fa277408afe : SUCCESS in 26m 30s","accounts_in_message":[],"_revision_number":7},{"id":"d1c97468292167b1c541957f98af004f5dd839b5","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-11-13 17:08:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96ff8afc25b0d5c5bcf722b1bc6fc6a69adfaa68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 18:09:20.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e8710a26d334ea4bae5b03c5d66fb01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630debdd302c4982bbfad38fac1676b3 : SUCCESS in 2m 55s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a6e3ad94c0d44265be188da22e2cd4b0 : SUCCESS in 31m 17s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c1ea51db6ee44ba18840ce17e8071f0b : FAILURE in 36m 48s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/edec482c1b1f42e5a49b14f65e74a1bc : SUCCESS in 33m 03s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/c71f03c32bfd41cebae8127980d6f352 : SUCCESS in 54m 24s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/caee64465ab046708941a0b5ecad9e15 : SUCCESS in 53m 30s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/ded91f09f7404d3bb342c02b248a9d54 : SUCCESS in 52m 51s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/d01b991753f64552aaa3da3ec8fe15f6 : SUCCESS in 48m 21s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/25a6d84d00624efe97dc8f89703bf298 : SUCCESS in 42m 37s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f4ec36e5ed2f409da2bff54403316d41 : SUCCESS in 44m 23s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/d34f697c2f074a4ebcb09cf2e172e689 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":7},{"id":"03db90334a549a2cb7d992eaf2d7e0216cc7639f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-01-17 13:54:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2da2d878c51a4b0ea463f939efb7bc40c3156b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 14:54:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dafc389511947348774134ddd44169c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb7b80562454a9faf0095a845ed5fe1 : SUCCESS in 2m 11s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7a8f9aa902724639a93e4802aaf4603f : SUCCESS in 50m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9962b609226046308288825264b749ee : FAILURE in 33m 08s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/0d611043573b4376a26690e1a748262a : SUCCESS in 29m 43s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/24d33ecade7547bc916cae8d6fcb25f8 : SUCCESS in 33m 54s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/d4af96631da5407c848b30762ef7cf08 : SUCCESS in 35m 02s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/5c8e130357754138904c28196533c008 : SUCCESS in 31m 33s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/b9416fdbfa134023b35c3f4f9d10ca86 : SUCCESS in 37m 14s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/569b25c26c7d4093a1a3ab49fdc4ffce : SUCCESS in 44m 58s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/e0e2236c2b234f8f965c3a3d7e26e1a8 : SUCCESS in 47m 23s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8b3c4002776a482daa5777d39ae83650 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":7},{"id":"b6e9470a0c4f44cb89dcff4e9be65470e930d371","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-01-18 15:39:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1e97a9153158ff006cc28f48fc461d09486c4b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:47:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cba391af03aa4cecb4207bbc12971b1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f629f3a900a94d0b918acea6e3b0175e : SUCCESS in 2m 15s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3b229498f66e4ea58ae5b9a4d8141714 : SUCCESS in 51m 52s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/22c077f55b9a4b74bf7418229443997d : SUCCESS in 49m 23s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/ea8ccb76a00b4ed5a25998f0c94e3525 : SUCCESS in 42m 42s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/0f4cdb0d443a4e0ebef00563b8fb554a : SUCCESS in 39m 34s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/7adae7fcd409421c8ccd9cf66417f841 : SUCCESS in 36m 36s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/21185b3f119b409eb65c617aa1ee491d : SUCCESS in 47m 18s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/073dd2b7d9e64e0e8f5497c69d2562ff : SUCCESS in 51m 01s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d9e17363628645f4ae0039cd17cb310f : SUCCESS in 45m 59s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/77f97f15786749febf958446a4c318c8 : SUCCESS in 29m 56s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/b62a5b2e10ce4a7d8d4fe4dec4e5eba0 : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":7},{"id":"afe90e08ef318d639ae8d366d17ffd5715f001d0","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-01-19 08:39:31.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e1915a2036764934521d46ba0ec4bd381e23fcd0","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-01-22 15:59:01.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"53fc6baf33766e183c92f1f8e1bf4111c078047e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-01-24 08:14:01.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ef87185480a287649db5012cec362f718e8712e1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-01-31 05:29:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3df70f3c6ef055b5f45de98a9dbd488bbe65aac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 05:30:02.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f481234ed4c18bb4b799810756ee7369d0d3db64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 05:55:03.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3d182aac2d4f958d97f5c27df6399f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a0230fe264453c8e1ac0ce704fe883 : SUCCESS in 3m 17s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ceb8503d03a742428d954b7acb56f65f : SUCCESS in 24m 59s","accounts_in_message":[],"_revision_number":7},{"id":"6f576d132ef9f129c8240e2957758372bdf01fd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 05:55:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"962ab9bb5311957b3304ccf3d25b9d8d0e15d095","revisions":{"c7ae6c94b689faf844acc8c8ec3b0438f5065adb":{"kind":"REWORK","_number":1,"created":"2023-10-25 14:56:20.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/1"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c7ae6c94b689faf844acc8c8ec3b0438f5065adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c7ae6c94b689faf844acc8c8ec3b0438f5065adb"}]},"branch":"refs/heads/master"},"3129e048690cd6c8b3934ff7eb452d553c3bb679":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-25 15:27:12.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/2"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-10-25 15:27:12.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin were missing idempotent IDs.\n\nThe \"check-uuid --fix\" command was implemented to identify and add the IDs to all test missing them.\n\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3129e048690cd6c8b3934ff7eb452d553c3bb679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3129e048690cd6c8b3934ff7eb452d553c3bb679"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0f04efbac2c134e99566027122de7a02b5b7f16":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-25 15:31:58.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/3"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-10-25 15:31:58.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin were missing idempotent IDs.\n\nThe \"check-uuid --fix\" command was implemented to identify all the tests that were missing idempotent IDs and added the IDs to the tests.\n\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c0f04efbac2c134e99566027122de7a02b5b7f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c0f04efbac2c134e99566027122de7a02b5b7f16"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6e29c7e3976fb826006619ed28b77a307ae44943":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-25 15:33:51.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/4"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-10-25 15:33:51.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin were missing idempotent IDs.\n\nThe \"check-uuid --fix\" command was implemented to identify all tests that were missing idempotent IDs and added the IDs to the tests.\n\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6e29c7e3976fb826006619ed28b77a307ae44943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6e29c7e3976fb826006619ed28b77a307ae44943"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1c70201f7727fad15bf2250500f0de90065ab59f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-11-08 13:56:43.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/5"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-11-08 13:56:43.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin were missing idempotent IDs.\n\nThe check-uuid tool was used to ensure that all tests have an id.\n\nCloses-Bug: #2030965\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1c70201f7727fad15bf2250500f0de90065ab59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1c70201f7727fad15bf2250500f0de90065ab59f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1e0de3b2bcf5174688184a0362a154d8e5279327":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-11-08 13:58:25.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/6"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-11-08 13:58:25.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin are missing idempotent IDs.\n\nThe check-uuid tool was used to ensure that all tests have an id.\n\nCloses-Bug: #2030965\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1e0de3b2bcf5174688184a0362a154d8e5279327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/1e0de3b2bcf5174688184a0362a154d8e5279327"}]},"branch":"refs/heads/master","description":"Edit commit message"},"962ab9bb5311957b3304ccf3d25b9d8d0e15d095":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-11-08 13:58:48.000000000","uploader":{"_account_id":36388,"name":"Jonathan Taymont","display_name":"Jon","email":"jonathan.taymont@gmail.com","username":"Jon"},"ref":"refs/changes/74/899274/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/74/899274/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/74/899274/7"}}},"commit":{"parents":[{"commit":"a34bd3b785a645b7879c7dd99625b7fa1546903b","subject":"Merge \"Pin barbican-tempest-plugin for jobs in EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a34bd3b785a645b7879c7dd99625b7fa1546903b"}]}],"author":{"name":"jont33","email":"jonathan.taymont@gmail.com","date":"2023-10-25 14:55:43.000000000","tz":120},"committer":{"name":"Jonathan Taymont","email":"jonathan.taymont@gmail.com","date":"2023-11-08 13:58:48.000000000","tz":0},"subject":"Barbican tempest missing IDs","message":"Barbican tempest missing IDs\n\nSeveral tests in the Barbican-tempest-plugin are missing idempotent IDs.\n\nThe check-uuid tool was used to ensure that all tests have an ID.\n\nCloses-Bug: #2030965\nChange-Id: Ice8a1c210e0ac2e50044f9a37e15b00fd9f306f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/962ab9bb5311957b3304ccf3d25b9d8d0e15d095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/962ab9bb5311957b3304ccf3d25b9d8d0e15d095"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
