)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~903679","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I9deaace9dcc804f517e59b36b9006dc9af2b84ab","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"fix/remove-multiple-secret-stores","hashtags":[],"change_id":"I9deaace9dcc804f517e59b36b9006dc9af2b84ab","subject":"Enable testing of multiple secret stores","status":"NEW","created":"2023-12-14 13:32:58.000000000","updated":"2023-12-18 11:26:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"7999837951121a0f61697323e44a36e52e267137","_number":903679,"virtual_id_number":903679,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-18 11:26:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-12-14 14:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7101637fbd995736b1ac07b4d7e567deb1fcbadf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-14 13:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"292debf1eaf98ac05c7f559434557b1c64a7152a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-14 13:42:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"965e0f9cd89ae0744ad901f5b19a689176ff789e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 14:27:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d60bd786c6c4ca99767873d08214898\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f409584796a147f29a820c0d825bfc8e : SUCCESS in 2m 27s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6140d888ee0b438fb7e25efd376d8ac6 : SUCCESS in 36m 14s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f338f836c5e14c7eb2067dff45052d7f : FAILURE in 43m 27s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/fa842aeda8da4858ba4f2cdab9c78481 : SUCCESS in 41m 20s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/2a1de6b476a34e7cac5e8af1c1201702 : SUCCESS in 40m 17s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/64b793486f60436c9417a19145bed492 : SUCCESS in 30m 57s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/a32e8518f61246d2a313a5052f6a4a6a : SUCCESS in 25m 13s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/dc0641607e15407b83bca032c9230c11 : SUCCESS in 29m 20s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/e21c00b3196e4effb4657d26b641ee88 : SUCCESS in 43m 21s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/7defc1342219488a80af216dbfd47772 : SUCCESS in 48m 21s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/66cf7cc40ff64fe391bbb480d56c1be1 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3b2fa4f60027bc65c7b3ae9528e3234a3fdc60dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-15 15:02:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4ab70d6bebe1f9ac0ede89cb7027799674bcb04f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-15 15:03:30.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"13c02c81b41aa71c8e6032c848b203e77f1e1ce7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-15 15:06:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c53d4d4ac325718965f87502f068dd6e09a34b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 16:02:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73d5cffa14a84853b85eb5242f3cf7c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61310116895a44f59aee34116f020393 : SUCCESS in 3m 09s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0a1cfd6e56ab4f83bf40211e1a6283b3 : SUCCESS in 39m 27s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/1bd86afa519541de9ffc482b36398b2b : FAILURE in 41m 50s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/b61e6d5aac174cd88236d5e2fd9013bd : SUCCESS in 43m 04s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/215a792a9c7d45db8174b3b1593217b3 : SUCCESS in 38m 56s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/6aaadf569037487eaee420c85127ce60 : SUCCESS in 28m 30s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/5178d93e0ffa4ebd842e0fd038129ee3 : SUCCESS in 28m 33s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/2921228f0aa04133bab0ba525ef229f9 : SUCCESS in 35m 14s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/bc68c4986ff24c7b8f0d9bda44d8cf1f : SUCCESS in 49m 10s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6a286eb59f024626b423d5026e6653c0 : SUCCESS in 44m 45s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/46344a4c81ec490aba4aff989cee9b1e : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":4},{"id":"1a2d87b440223b179006a2fd8a1d8d62f8803270","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-15 16:20:11.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8fb94878a2f73d774ca986e01bf62bac9ec369b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 17:19:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb43ed10f0e64893bc6d59262de58749\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292aacec13914a8f851f8dfb9be516c0 : SUCCESS in 2m 21s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/aa49c720c617441591b73c61d5910be8 : SUCCESS in 52m 29s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/236c804dc56f4a3bab3efbfab1f003c0 : SUCCESS in 36m 26s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/1afb3d4d4f0e44269292bd607470efdc : SUCCESS in 30m 55s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/bdc43fb873f0478e86f3f88e4925623e : SUCCESS in 40m 58s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/ed9e751d84674a60b20727b3f701bdc6 : SUCCESS in 47m 09s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/333c446d9eb6437eaf08b40be47c1d26 : SUCCESS in 37m 21s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/d598c3f883d84d3b8ae874d3846fe1eb : SUCCESS in 52m 31s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6c3f2b3848054eb184c7cf8bf0644ec3 : SUCCESS in 31m 39s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/15e8894d136c4b55bbe9621541170cd3 : SUCCESS in 34m 13s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8ea132475a6640e89d57d9408b9a65e5 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":5},{"id":"f026b61a4534227f7bc858a8dd4f9929b1da674d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-18 10:32:24.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7999837951121a0f61697323e44a36e52e267137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 11:26:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fd46b37196494bba63dd599d5374a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9852472c5464d1e9a4041bacca1bf46 : SUCCESS in 2m 37s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9eacb5af5600478183d14a2238b3fe07 : SUCCESS in 29m 56s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/fa58f5bd525e4d13a78b18cbc08eaea2 : SUCCESS in 36m 11s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/c318d352749d41af98d635362ed9fb60 : SUCCESS in 47m 55s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/44122681a0794e34976bb7c5c5219340 : SUCCESS in 44m 38s\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/313f1ec585a1482d9de0f88b4c2d09a9 : SUCCESS in 41m 51s\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/fb5975e95fcf4e35b45e35f96319c609 : SUCCESS in 40m 32s\n- barbican-tempest-plugin-simple-crypto-wallaby https://zuul.opendev.org/t/openstack/build/16f921dec45a45e48592ed36120b7547 : SUCCESS in 39m 43s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d1fb659ae54741ed8d8709bbd1bafe5a : SUCCESS in 32m 50s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/5b4525a491914d86b2ff5bed9799b60a : SUCCESS in 40m 50s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/bd202e353be84bf5947e9003b1b43bd5 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f19f2d239ba455580faa3a53b5b26ad9e8184df2","revisions":{"022a9610ca2169f5e36951ed6781a52114e3cd04":{"kind":"REWORK","_number":1,"created":"2023-12-14 13:32:58.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/1"}}},"commit":{"parents":[{"commit":"193a7eeb0a40dc5fe26cdbdd1d40f04ec20c0fb1","subject":"Pause testing of multiple secret stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/193a7eeb0a40dc5fe26cdbdd1d40f04ec20c0fb1"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/022a9610ca2169f5e36951ed6781a52114e3cd04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/022a9610ca2169f5e36951ed6781a52114e3cd04"}]},"branch":"refs/heads/master"},"0f34ba20b467b8366af31262e156c07fe6dfe838":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-15 15:02:45.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/2"}}},"commit":{"parents":[{"commit":"06af350a630569708b2ae17158004482b1538520","subject":"Update tests for secret:delete and secret:get policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/06af350a630569708b2ae17158004482b1538520"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-15 15:02:23.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0f34ba20b467b8366af31262e156c07fe6dfe838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0f34ba20b467b8366af31262e156c07fe6dfe838"}]},"branch":"refs/heads/master"},"9cb580ffae9474c50e96a2d58d771a59c8bd54af":{"kind":"NO_CHANGE","_number":3,"created":"2023-12-15 15:03:30.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/3"}}},"commit":{"parents":[{"commit":"0b3d0268212809b5044ca993673789c841c70e15","subject":"Pause testing of multiple secret stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0b3d0268212809b5044ca993673789c841c70e15"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-15 15:03:18.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9cb580ffae9474c50e96a2d58d771a59c8bd54af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9cb580ffae9474c50e96a2d58d771a59c8bd54af"}]},"branch":"refs/heads/master"},"af355b38d1d74a4aadd475194aaf52341a8bea4b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-15 15:06:15.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/4"}}},"commit":{"parents":[{"commit":"b3efd9999f25d1d96fd00563fada0ac33250f47c","subject":"Pause testing of multiple secret stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b3efd9999f25d1d96fd00563fada0ac33250f47c"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-15 15:06:03.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/af355b38d1d74a4aadd475194aaf52341a8bea4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/af355b38d1d74a4aadd475194aaf52341a8bea4b"}]},"branch":"refs/heads/master"},"190832182b9b16b3d943fa1fba180847fd06dbad":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-15 16:20:11.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/5"}}},"commit":{"parents":[{"commit":"baf3f1c094c579096c95ca9e6645cd0cd9e9d125","subject":"Pause testing of multiple secret stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/baf3f1c094c579096c95ca9e6645cd0cd9e9d125"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-15 16:19:48.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/190832182b9b16b3d943fa1fba180847fd06dbad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/190832182b9b16b3d943fa1fba180847fd06dbad"}]},"branch":"refs/heads/master"},"f19f2d239ba455580faa3a53b5b26ad9e8184df2":{"kind":"NO_CHANGE","_number":6,"created":"2023-12-18 10:32:24.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/79/903679/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/79/903679/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/79/903679/6"}}},"commit":{"parents":[{"commit":"6345b34ae2f47346dd72dc196ec89a3793f57eac","subject":"Pause testing of multiple secret stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6345b34ae2f47346dd72dc196ec89a3793f57eac"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-14 13:31:41.000000000","tz":60},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-12-18 10:32:14.000000000","tz":60},"subject":"Enable testing of multiple secret stores","message":"Enable testing of multiple secret stores\n\nThe issue with multiple secret stores is resolved [1]. We can enable\nthe testing of this feature in the SRBAC job again.\n\nRelated-Bug: #2043457\nChange-Id: I9deaace9dcc804f517e59b36b9006dc9af2b84ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f19f2d239ba455580faa3a53b5b26ad9e8184df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f19f2d239ba455580faa3a53b5b26ad9e8184df2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
