)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~916887","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I6d46105b8f44e279cdae1b739d868f1ef40689a1","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"bz2026029","attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-05-22 07:50:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-05-22 08:39:18.000000000","reason":"A robot voted negatively on a label"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2024-04-25 22:19:24.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6d46105b8f44e279cdae1b739d868f1ef40689a1","subject":"Add a new test to delete someone elses volume","status":"NEW","created":"2024-04-24 11:05:38.000000000","updated":"2024-07-24 12:53:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4ec7db767553f658e71bc6bbc8c550a1ad516c3","_number":916887,"virtual_id_number":916887,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-30 22:21:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":2,"date":"2024-07-24 12:53:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 11:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-24 14:14:17.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2024-05-14 06:14:31.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2024-07-24 12:53:46.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"f9304d1e39e57a92634b9497c85f039a2d07054a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-24 11:05:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d8fbf22e7dac8818d78135166bd58b29c43d0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 11:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957118fde64548318f24a5a7bc01a13e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9badc82601b647fdb2d57062aeeec181 : SUCCESS in 2m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c3432af8f6054a01b3dc959a011383cf : SUCCESS in 29m 57s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/a347b4da96144b1288df6bd9a6d34d96 : SUCCESS in 27m 18s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/319a2af1dc9a4ef69c241d7d70416539 : SUCCESS in 26m 15s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/bf75efc9e9224915841e51fb7f762d65 : SUCCESS in 43m 11s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/63e2c65a57fc4237aedb84afb50e967a : SUCCESS in 30m 33s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/7ea84c57ed564190a7b69bc951cbb089 : SUCCESS in 36m 34s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/5a3ceecf4b804c2ebe8d558bf0bece13 : SUCCESS in 40m 17s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/43c49c14d74f4807bf98228f362cf938 : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"54b0c131859e79919f0da044a9d0e93c87255d7f","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2024-04-24 14:14:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7965a65737f9cba764e99eb912c81f7fa53b51d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-25 22:19:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee38235f78d107c0f6e7e6f8adab7a7b9b8666e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 23:02:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6410628ca7b54808a66a202ca2ba849f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8411b54092784b92bbe85b4c3e964a68 : FAILURE in 2m 29s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/43bf190a088c4572a6109e731de4f1d6 : SUCCESS in 37m 17s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/631bd8eea92c4f32b9cdd48c1ab4ff87 : SUCCESS in 29m 21s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/7d586ea9311044a988ca2d43078a8499 : SUCCESS in 37m 01s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/ba28e0781e2248e286da0f2824e2e5b6 : SUCCESS in 23m 37s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/b7e4cb4d327142da9fe8468082f4a635 : SUCCESS in 24m 08s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/0547e438c401462ab898bd7b810412a0 : SUCCESS in 25m 11s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6d933d3c02e945ef8ffc5b2281bb4a8e : SUCCESS in 24m 18s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/11c27a15dc0e4347bc4b05a652cdbe66 : SUCCESS in 26m 48s\n\nWarning:\n  Comments left for invalid file barbican_tempest_plugin/tests/scenario/barbican_manager.py","accounts_in_message":[],"_revision_number":2},{"id":"4ff931245f2360aae6fc35ee8b4272eaaa7ece94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-03 07:40:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a8016f67b5bfbde5b970cf3a8500307dcebcc2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 08:35:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b1fc7286f942ab9a8be5cf4adb9919\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9941ee5d60747479b28d2ec7f79a922 : SUCCESS in 2m 02s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4741f378408f416fac2f9b01b11b898e : SUCCESS in 45m 01s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/479fcc4f453f42278a4cb8cfe6878119 : SUCCESS in 39m 53s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/3df8297110d040bb977a108099a3848c : SUCCESS in 31m 44s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/05e0b76ff3354fe8852dce3d298957bd : SUCCESS in 35m 44s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/dd6d91ccfb8d42718dfa71f4fd9dd385 : SUCCESS in 24m 12s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/87d8c9e8a66a431897bd7f4c628702d3 : SUCCESS in 37m 23s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ddbf6a67ccb84cde9faa28f856d372da : SUCCESS in 31m 05s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8303846547784219887ce7da737cae3f : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ca0742685aceda194c26af972712bfa677c05723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-03 09:03:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c02c9baa3cfbd9547bc0fa8957a12e3c9338c3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 09:58:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a850edb3854c978dbb5a408c9642b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8ac79d019b4000b0cc17d0c221da0d : SUCCESS in 2m 02s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/76c2c70bdfd0476a8ebe5a41f9a49c64 : SUCCESS in 38m 55s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/a038e56279f4458a9d6db4afe3f93f94 : SUCCESS in 39m 12s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/a79ef6fa5e114428b0b8647e0773bbf2 : SUCCESS in 35m 19s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/fae81394b6d644dfaecaddafa838df70 : SUCCESS in 35m 33s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/4f7737dc755e4143983bb71ff94b4358 : SUCCESS in 40m 05s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/391927e42d044cf3beb451ff1a260f3b : SUCCESS in 38m 20s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b1299dd3de02459daf78dd3fed7cbbaf : SUCCESS in 26m 19s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/67b6e52a0c3d4b6db9252f5359d94db0 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":4},{"id":"c4b7278d29e1d2b07387a29c089ea2f0c610ce57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-06 06:36:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"207fbef4d5ab70e90cdcc73601e77bcb394b278b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 07:22:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b41d94828bb463784a534319892bb73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e0f4bfdb104ccfae387ac6b76442fb : SUCCESS in 2m 38s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/caef8c7be12f42a6a252bbfc026440e3 : SUCCESS in 26m 33s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/33a66bb61bf04709840283f02ce8ca07 : SUCCESS in 29m 24s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/a9ebd9902854475eb293e5bab2575e27 : SUCCESS in 30m 56s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/3e18b6f34c214190976b61139127820b : SUCCESS in 40m 23s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/d64398aa062e4acd9aa050e80f723763 : FAILURE in 20m 06s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ea985480cd3e454c931463738d48ba0e : SUCCESS in 38m 04s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/99ec3b1cc0d94d35865515dff431cb5e : SUCCESS in 39m 51s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/480ab729186040f59ffc8bc6f4b4fab6 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e4d7dc894fca6c4b8d7319435f2648194a103aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-07 07:26:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ebd83ad4b91f79331a33db317be682e023361bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 08:15:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d58ff9ddd0fc4dbd9f2f50ae068a0ea2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87c3ed19b784dcdb1295d31df8ffe45 : SUCCESS in 2m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fc6fbf9b4f56459b9f83a9f5bb95e978 : SUCCESS in 37m 24s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/54c53e0abdbf44e58997ebb0fee2304d : SUCCESS in 34m 58s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/f7fc232124da42728eb0474de762b24f : SUCCESS in 36m 13s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/bed5dd1101c54d5d860a4c7a3af7d412 : SUCCESS in 36m 41s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/a87459fab9094b4aae4d1a28e60b839f : FAILURE in 29m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a55b68526146417990df62441fa26079 : SUCCESS in 41m 08s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/3e7775e0afb6470bb054c5a7587ea0a7 : SUCCESS in 25m 26s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/dc09a63a0a234713bedf126b7a9e8d9b : SUCCESS in 25m 36s","accounts_in_message":[],"_revision_number":6},{"id":"37ad4ad55df1b80fff22792a83db3c60b0968dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-07 10:38:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"74e52999ae71dd7b304656732b27262a00c16ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 11:19:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9620ad7fa4d4c3db812eaf9c2fc7d7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889233331dc24dadb32ccd5f64b4620a : SUCCESS in 2m 00s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/752d83c75b904118b4adf459fe142071 : SUCCESS in 35m 20s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c7e67dc2f43c4c7bbbdfceca8aa5ae01 : SUCCESS in 40m 08s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/acc59c529b62448fa3b1777c51ed839c : SUCCESS in 26m 21s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/64d6562c9109450fbb52bd55e20baf71 : SUCCESS in 27m 29s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/aa1bab335f5546738e02bd54c12b0b9a : FAILURE in 16m 55s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/1927bb4f978f40ed9c9f4021f3e0e495 : SUCCESS in 25m 44s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/260ed5022603425abe1cf802cbf0421a : SUCCESS in 31m 16s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/2068345ad5ef4045829355f11d48eddd : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":7},{"id":"8420ced13045076b8925d4d24ce1542d700d16c9","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-05-14 06:14:31.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6ad0d585681c0642b3c013632c67d96872a63729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-14 09:22:52.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ce3a9e902bab80593b4b8627adf9e946659e1673","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-14 09:23:13.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cf540a2bedd0a1603b4c78e522ad24369c7095fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:13:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cba9766a58241c5a3927c0ebd09cd19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88647d07c4de4f75abcf9f02a1e05df6 : SUCCESS in 2m 22s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3d1ae373d59d43389eb6a531181e1a86 : SUCCESS in 24m 10s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/257bced729f641c0848542fe00a6b622 : SUCCESS in 40m 03s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/513d0e29531e4f2d9ac20d177901e82a : SUCCESS in 25m 51s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/b4f43a0a135f4f83bdc7e710165617b9 : SUCCESS in 24m 27s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/1dc2cdf4d69a44ac80947a56c1cb182e : FAILURE in 5m 52s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/79ce4885e7c34226b30b212137eef28c : SUCCESS in 27m 02s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/32296e4351c54260b9c219b347250982 : SUCCESS in 26m 47s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a01454a02bcb440e849f8afa60ccf59b : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":8},{"id":"29771806f331ea3bf75809cff949ef50fd5c6592","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-05-14 15:10:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"af068526ba0da47fe25f3cc9199ca5aca3773e3a","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-05-14 15:14:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3ba2ce6443bb79fb98b5cf57fa3635236d3e5032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-22 07:50:07.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"178e99cc3e6635de48607a10bfdd76544840f831","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-22 07:50:12.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4c29685e358f5dce3b59bf874ef8085fe1e16aaa","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-22 08:01:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e037c64df891b05019eadd652abb67a037a4ded7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 08:39:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5c30e704dfb4320a2d0d845accc316d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af7761ed78245a8853204c64df40456 : SUCCESS in 2m 26s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e6becf2bb9df4056825e81115f31c204 : SUCCESS in 25m 22s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/425a7bd0364a4db0acd4fc875895046f : SUCCESS in 27m 20s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/40fdc4610eb84068aa9a495d4924b933 : SUCCESS in 22m 43s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/b72bf19b4fec4071a25786d51d910326 : SUCCESS in 38m 41s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/2c5365778ace4a6ebe40738a734883b0 : FAILURE in 4m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6e4619a423344571b296a1d01501b153 : SUCCESS in 37m 54s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f56bc89550c84a989d4944ea741ef4c7 : SUCCESS in 27m 02s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/7654e885391048359808234c4de97b1b : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":9},{"id":"210471b905d535b5027c2290b3b2ac336a12fea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-06-30 17:01:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"525da8a854609926a343c6d312a8cee71f355d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 17:39:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aa64b2078ce4dfda6fba838a41d83a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547fc06ce33240aba3b1338064b34155 : SUCCESS in 5m 17s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/70af427e182e41cc826c804109cca0c1 : SUCCESS in 26m 05s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f60c2238c0014154bdec13cc31d8c2ef : SUCCESS in 33m 12s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/675b4ea1c31243d8be447fe7fc9dac54 : SUCCESS in 22m 04s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/4ec0fe6e4768448ea0688d0ab1984f5f : SUCCESS in 28m 59s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/eea0d158d88c4a4ab7ab17f2b4b1cdec : SUCCESS in 37m 43s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/4331c7ad7550460f8c53529ca617cf16 : FAILURE in 13m 09s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/2d43a3197e274bf1af2ac0e27b61c64d : SUCCESS in 25m 57s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/6b0861a362f144198a58f170b6c65188 : SUCCESS in 19m 59s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/e526177151ee44bca136474e39a8427d : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":10},{"id":"55c447e6f601448b71d5ca7c7871abd45d5379bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-06-30 21:30:41.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6374bc5fc48d002081769eaeae1627b686a86be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 22:21:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3901c92930114c058e7dc41d02af2cb9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6931d947a7d14bd3b459c9b898235f15 : SUCCESS in 6m 22s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2974274d4f064f768c3e63061031d0d8 : SUCCESS in 34m 24s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/34949494da9a4de9a25fba5ddfdb76ea : SUCCESS in 21m 47s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/48320f44e26744eeacf86997ff876875 : SUCCESS in 23m 54s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/94fb9c29ce144d55a398dc760a2802b4 : SUCCESS in 36m 30s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/78f5e2013c584f7c8a1a5c37d67d7c01 : SUCCESS in 38m 15s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/24656517d01544159c9384c9af197406 : SUCCESS in 31m 08s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d70a7c4336f04573af40119591aa7b5e : SUCCESS in 33m 25s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f4883b0829744d29a3facbc61daf566d : SUCCESS in 36m 02s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a8944fc62ef84a87a4567a463a17dc5b : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":11},{"id":"b4ec7db767553f658e71bc6bbc8c550a1ad516c3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-24 12:53:46.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"dec5cfb30762542277979d59d4725d4e0cfd7bbc","revisions":{"5aa037a59901b3199b9e684dd73e7b7b709303af":{"kind":"REWORK","_number":1,"created":"2024-04-24 11:05:38.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/1"}}},"commit":{"parents":[{"commit":"a7f6223de4036bc6a27e3bc365b4abfbd915595b","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a7f6223de4036bc6a27e3bc365b4abfbd915595b"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5aa037a59901b3199b9e684dd73e7b7b709303af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5aa037a59901b3199b9e684dd73e7b7b709303af"}]},"branch":"refs/heads/master"},"46d8c3016aa430e101b9b53b8736571be410b8fe":{"kind":"REWORK","_number":2,"created":"2024-04-25 22:19:24.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/2"}}},"commit":{"parents":[{"commit":"c5eab321b48001e0eeaff1a26b0a68d7b39ddc8d","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c5eab321b48001e0eeaff1a26b0a68d7b39ddc8d"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 22:18:51.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/46d8c3016aa430e101b9b53b8736571be410b8fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/46d8c3016aa430e101b9b53b8736571be410b8fe"}]},"branch":"refs/heads/master"},"258fd943d34459e2cabd74e854a6059be292266f":{"kind":"REWORK","_number":3,"created":"2024-05-03 07:40:18.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/3"}}},"commit":{"parents":[{"commit":"7d400b5df8cf89ee3dc9a59a1d49a6c8ea54ebbe","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7d400b5df8cf89ee3dc9a59a1d49a6c8ea54ebbe"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 07:32:58.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/258fd943d34459e2cabd74e854a6059be292266f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/258fd943d34459e2cabd74e854a6059be292266f"}]},"branch":"refs/heads/master"},"f61ff25339df728fd22b105ec424f5f0d728f358":{"kind":"REWORK","_number":4,"created":"2024-05-03 09:03:43.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/4"}}},"commit":{"parents":[{"commit":"a91a74c95eef34c38e94c93ce5e361341741b38e","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a91a74c95eef34c38e94c93ce5e361341741b38e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 09:02:30.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f61ff25339df728fd22b105ec424f5f0d728f358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f61ff25339df728fd22b105ec424f5f0d728f358"}]},"branch":"refs/heads/master"},"5083886f34bfc173c06e206b782e1c3e0e790e34":{"kind":"REWORK","_number":5,"created":"2024-05-06 06:36:18.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/5"}}},"commit":{"parents":[{"commit":"a91a74c95eef34c38e94c93ce5e361341741b38e","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a91a74c95eef34c38e94c93ce5e361341741b38e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-06 06:35:34.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5083886f34bfc173c06e206b782e1c3e0e790e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5083886f34bfc173c06e206b782e1c3e0e790e34"}]},"branch":"refs/heads/master"},"b8dae9309d181410a009626a7074027a93506b09":{"kind":"REWORK","_number":6,"created":"2024-05-07 07:26:01.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/6"}}},"commit":{"parents":[{"commit":"a91a74c95eef34c38e94c93ce5e361341741b38e","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a91a74c95eef34c38e94c93ce5e361341741b38e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-07 07:25:33.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b8dae9309d181410a009626a7074027a93506b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b8dae9309d181410a009626a7074027a93506b09"}]},"branch":"refs/heads/master"},"cec7bcf190eb07bc37a31e01ce49ed7bf8b6aa0e":{"kind":"REWORK","_number":7,"created":"2024-05-07 10:38:53.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/7"}}},"commit":{"parents":[{"commit":"a91a74c95eef34c38e94c93ce5e361341741b38e","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a91a74c95eef34c38e94c93ce5e361341741b38e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-07 10:38:38.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/cec7bcf190eb07bc37a31e01ce49ed7bf8b6aa0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/cec7bcf190eb07bc37a31e01ce49ed7bf8b6aa0e"}]},"branch":"refs/heads/master"},"c74415bc8fbf4862b24af027508deea78a14d149":{"kind":"REWORK","_number":8,"created":"2024-05-14 09:22:52.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/8"}}},"commit":{"parents":[{"commit":"9d2693745d33adfa17d877db32ccf97c298057c3","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9d2693745d33adfa17d877db32ccf97c298057c3"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-14 09:15:18.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c74415bc8fbf4862b24af027508deea78a14d149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c74415bc8fbf4862b24af027508deea78a14d149"}]},"branch":"refs/heads/master"},"255f44270828faa3168af1f464cd42c7df39ef44":{"kind":"REWORK","_number":9,"created":"2024-05-22 07:50:07.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/9"}}},"commit":{"parents":[{"commit":"9d2693745d33adfa17d877db32ccf97c298057c3","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9d2693745d33adfa17d877db32ccf97c298057c3"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-22 07:49:01.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/255f44270828faa3168af1f464cd42c7df39ef44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/255f44270828faa3168af1f464cd42c7df39ef44"}]},"branch":"refs/heads/master"},"9604ac7226717ec0fb51a048c3962f590b22160f":{"kind":"REWORK","_number":10,"created":"2024-06-30 17:01:02.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/10"}}},"commit":{"parents":[{"commit":"39e7526e81c7f0aa80bd9259dbf4727355cd672f","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/39e7526e81c7f0aa80bd9259dbf4727355cd672f"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-06-30 16:51:05.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9604ac7226717ec0fb51a048c3962f590b22160f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/9604ac7226717ec0fb51a048c3962f590b22160f"}]},"branch":"refs/heads/master"},"dec5cfb30762542277979d59d4725d4e0cfd7bbc":{"kind":"REWORK","_number":11,"created":"2024-06-30 21:30:41.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/87/916887/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/87/916887/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/87/916887/11"}}},"commit":{"parents":[{"commit":"302abc79561ca33b7b2a23eb9bd69fd6b24a0114","subject":"Introduce a new test for secondary encrypted volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/302abc79561ca33b7b2a23eb9bd69fd6b24a0114"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-24 11:04:11.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-06-30 21:30:09.000000000","tz":120},"subject":"Add a new test to delete someone elses volume","message":"Add a new test to delete someone elses volume\n\nThe test case verifies that an admin user can delete other user\u0027s\nencrypted volume after the user and the project have been deleted.\n\nChange-Id: I6d46105b8f44e279cdae1b739d868f1ef40689a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/dec5cfb30762542277979d59d4725d4e0cfd7bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/dec5cfb30762542277979d59d4725d4e0cfd7bbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
