)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~916969","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-04-25 06:24:31.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13","subject":"Add rocky job and add new Python version","status":"NEW","created":"2024-04-25 06:24:04.000000000","updated":"2024-05-28 06:59:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ccf7e650360e171dfb7a8f20325e2ce1eb69178d","_number":916969,"virtual_id_number":916969,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-28 06:59:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 06:24:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58d5b870be174447c1f341d1190a1aaa1c48ff34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-25 06:24:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82645f4ffa3b959c91892eb547483fa2c56c7eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 06:24:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/barbican-tempest-plugin on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/barbican-tempest-plugin/.zuul.yaml@master\", line 157, column 11","accounts_in_message":[],"_revision_number":1},{"id":"c7a4d97356f353e2b5dc42e19f74ddd72d032857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-25 06:37:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8a7a73bb4f5b47c2a9ada73704fd4ac6fd61cfed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 06:38:15.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/barbican on branch unmaintained/wallaby.  The\nproblem was:\n\n  Job barbican-tempest-plugin-simple-crypto-wallaby not defined\n\nThe problem appears in the the \"barbican-simple-crypto-devstack-tempest\" job stanza:\n\n  job:\n      name: barbican-simple-crypto-devstack-tempest\n      parent: barbican-tempest-plugin-simple-crypto-wallaby\n      description: |\n        Compatibility alias for barbican-tempest-plugin-simple-crypto\n\n  in \"openstack/barbican/.zuul.yaml@unmaintained/wallaby\", line 53","accounts_in_message":[],"_revision_number":2},{"id":"36e5647eee61d9de3e7739b11846f6982958c012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-25 07:14:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c6a5419243319f379945e11d641e572a42446b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 08:00:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852fcc7b95e9451f92f0c2691db521d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9356b14342644bd0b858614c34a62a0d : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/357eb6243b8849e182da47409ef5bbec : SUCCESS in 2m 32s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c99a5527b762483fbc8cf62727597ab5 : SUCCESS in 35m 47s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/181c5ea668904f59a59a4f6d2656adda : SUCCESS in 28m 03s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/7384e958d853403bb5907d12f6caf547 : SUCCESS in 25m 42s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/b27fe14321ae4df3a6eadc664558c85c : SUCCESS in 35m 13s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/a9c50506602141219e064cc351a15bbf : SUCCESS in 39m 40s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/346bb3b24ab8488b8a48c2becab2454c : SUCCESS in 28m 12s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/994baddc5eb84f44a461c58f1172e1e9 : SUCCESS in 37m 37s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b65e9856d3e040038ace29332308d5fb : SUCCESS in 38m 23s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/63e38515f20a41a78c68a3568ee87a36 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b141afed609a2e0f4d3e663098ce99e6d39862b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-02 14:20:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"831ad4c061abffbaae98f6dd1c92c1c3d4558506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:18:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d788b43eccda4264843ae39a83a1803c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/33f83e52e6b14c90a71efc91521669b2 : SUCCESS in 1m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25daba01103748e4b687578cd7c45a64 : SUCCESS in 2m 26s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fd83c7c5f53447b38244cf17c58d9f62 : SUCCESS in 34m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/353da86bf613429aaf34930ab0a126d0 : SUCCESS in 41m 52s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/ba693d0f616b4aed807fcbedc13a4a9a : SUCCESS in 38m 44s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/ea4695f4d05049549793597190491dc7 : SUCCESS in 39m 05s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/e5c3e3906f7b423fb05b4bfc0f1e7084 : SUCCESS in 40m 25s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/9978e8fdfe8b44cda6cffe800411cd57 : SUCCESS in 41m 32s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/edfa7f0929d44bf79667582d72a1ec4e : SUCCESS in 35m 25s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/d50a4b0deeee41ed80b79e8451264a89 : SUCCESS in 40m 34s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f971e5edc87a47a3946c9d60da511a33 : SUCCESS in 29m 23s\n- barbican-tempest-plugin-simple-crypto-rocky https://zuul.opendev.org/t/openstack/build/bcb6c522106e4cc4a05ff47b8f16688a : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c7eb1318e0db2c878bad8c0051cdd6787ae5e38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-28 06:11:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ccf7e650360e171dfb7a8f20325e2ce1eb69178d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 06:59:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3ef6a6ffd6142e5aff857246683e14b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be6c4c59c2d644c485ff6cab792110f5 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2376885aa97d43249b2184b632981f13 : SUCCESS in 2m 18s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1fb119484142458cb78dd7577b98545d : SUCCESS in 23m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9c81222da1944b278d22d133fdb74be2 : SUCCESS in 21m 43s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/5ee40959251e4853890270f52d882365 : SUCCESS in 22m 13s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/1e30e3dd31df4b6ca9148869abd61454 : SUCCESS in 20m 24s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/0748c6789d4646fd9626cadf2a0798d0 : SUCCESS in 38m 35s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/3971ab51cd63495fbf4d4849bba2a6cd : FAILURE in 13m 41s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d8829bae205e4f3280f3ea074576ef44 : SUCCESS in 25m 48s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b9c1ed951c9f4483a8a335d60457b111 : SUCCESS in 38m 58s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/965811fc795d412ea17a0f92764ce288 : SUCCESS in 24m 30s\n- barbican-tempest-plugin-simple-crypto-rocky https://zuul.opendev.org/t/openstack/build/f7494a61b08e476283d3346cae43e55f : FAILURE in 32m 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"01048d9b3bfdc816ba2a877284c9aad4d6c5c5cf","revisions":{"0bbe6581be8eab51362cb0e9718c3bd39299796c":{"kind":"REWORK","_number":1,"created":"2024-04-25 06:24:04.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/69/916969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/69/916969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/1"}}},"commit":{"parents":[{"commit":"82b0e48639ad52838902dffbc3acaa5bc789b87e","subject":"Merge \"Support files removed for CONF.scenario.img_dir option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82b0e48639ad52838902dffbc3acaa5bc789b87e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"subject":"Update jobs for the current cycle","message":"Update jobs for the current cycle\n\nThis patch removes all jobs that are not referenced anywhere and\nadds a new Python version to setup.cfg.\nApart from that this adds 2 new jobs, one on stable/2024.1 and one\non master but on rockylinux.\n\nChange-Id: Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0bbe6581be8eab51362cb0e9718c3bd39299796c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/0bbe6581be8eab51362cb0e9718c3bd39299796c"}]},"branch":"refs/heads/master"},"fe2e392c107643ea080e3c125ff993249b6878d2":{"kind":"REWORK","_number":2,"created":"2024-04-25 06:37:55.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/69/916969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/69/916969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/2"}}},"commit":{"parents":[{"commit":"82b0e48639ad52838902dffbc3acaa5bc789b87e","subject":"Merge \"Support files removed for CONF.scenario.img_dir option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82b0e48639ad52838902dffbc3acaa5bc789b87e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:37:49.000000000","tz":120},"subject":"Update jobs for the current cycle","message":"Update jobs for the current cycle\n\nThis patch removes all jobs that are not referenced anywhere and\nadds a new Python version to setup.cfg.\nApart from that this adds 2 new jobs, one on stable/2024.1 and one\non master but on rockylinux.\n\nChange-Id: Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fe2e392c107643ea080e3c125ff993249b6878d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/fe2e392c107643ea080e3c125ff993249b6878d2"}]},"branch":"refs/heads/master"},"a73646a2669b6fd538975686290b819e3132bd24":{"kind":"REWORK","_number":3,"created":"2024-04-25 07:14:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/69/916969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/69/916969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/3"}}},"commit":{"parents":[{"commit":"82b0e48639ad52838902dffbc3acaa5bc789b87e","subject":"Merge \"Support files removed for CONF.scenario.img_dir option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82b0e48639ad52838902dffbc3acaa5bc789b87e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 07:14:08.000000000","tz":120},"subject":"Update jobs for the current cycle","message":"Update jobs for the current cycle\n\nThis patch adds a new Python version to setup.cfg and adds 2 new\njobs, one on stable/2024.1 and one on master but on rockylinux.\n\nChange-Id: Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a73646a2669b6fd538975686290b819e3132bd24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/a73646a2669b6fd538975686290b819e3132bd24"}]},"branch":"refs/heads/master"},"43caa90930c4e91aadfe124e3f97c293d6815a4b":{"kind":"REWORK","_number":4,"created":"2024-05-02 14:20:30.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/69/916969/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/69/916969/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/4"}}},"commit":{"parents":[{"commit":"82b0e48639ad52838902dffbc3acaa5bc789b87e","subject":"Merge \"Support files removed for CONF.scenario.img_dir option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82b0e48639ad52838902dffbc3acaa5bc789b87e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-02 14:19:47.000000000","tz":120},"subject":"Update jobs for the current cycle","message":"Update jobs for the current cycle\n\nThis patch adds a new Python version to setup.cfg and adds 2 new\njobs, one on stable/2024.1 and one on master but on rockylinux.\n\nChange-Id: Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/43caa90930c4e91aadfe124e3f97c293d6815a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/43caa90930c4e91aadfe124e3f97c293d6815a4b"}]},"branch":"refs/heads/master"},"01048d9b3bfdc816ba2a877284c9aad4d6c5c5cf":{"kind":"REWORK","_number":5,"created":"2024-05-28 06:11:19.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/69/916969/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/69/916969/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/69/916969/5"}}},"commit":{"parents":[{"commit":"d7f96744c352e16301832751ae4af14818c4ec5c","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d7f96744c352e16301832751ae4af14818c4ec5c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-04-25 06:21:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-28 06:09:52.000000000","tz":120},"subject":"Add rocky job and add new Python version","message":"Add rocky job and add new Python version\n\nThis patch adds a new Python version to setup.cfg and\none job on master using rockylinux.\n\nChange-Id: Ibf6e1a9e11f3f954d7d6218fbfd5b44dfe488e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01048d9b3bfdc816ba2a877284c9aad4d6c5c5cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/01048d9b3bfdc816ba2a877284c9aad4d6c5c5cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
