)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~932946","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I86b6abfdbd06851f6d1b331ce113e0897384bd28","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-11-18 16:58:50.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-30 18:45:53.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I86b6abfdbd06851f6d1b331ce113e0897384bd28","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-10-22 01:30:36.000000000","updated":"2024-11-30 20:11:40.000000000","submitted":"2024-11-30 20:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932946-migrate-to-noble","meta_rev_id":"0a82515b10c6181d8ba5ac4d62f55c5c6efc800d","_number":932946,"virtual_id_number":932946,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-30 20:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-11-18 16:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-11-18 16:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 02:20:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 10:25:24.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-11-18 16:19:14.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"531c91937c9a5a4223a28a886f206dbde4c4ac00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 01:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b349564a9d81cd550570b8c87a755784ee621b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 02:20:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8c83828f81a432cb8abd75dc37d6de3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f79d94ce9e346279a0641fbd2a4f376 : FAILURE in 3m 12s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d19df9b11ab4e13ad601881a341bad8 : SUCCESS in 24m 03s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/e56ba36a52c949e1a9750a21ac9c1c47 : SUCCESS in 18m 07s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/8dcd1fc02b9443a6bd9084ff38c4792d : SUCCESS in 40m 20s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/deb389fd0a314328ac07548a4bbd39d9 : SUCCESS in 24m 05s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/ef871b6903a747f78571ba602f4747fa : SUCCESS in 20m 45s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/a4777115388f4a519daa1154d9956ace : SUCCESS in 40m 31s\n- barbican-tempest-plugin-simple-crypto-zed https://zuul.opendev.org/t/openstack/build/ef10baebfc0248f6844bf6ec40a1a396 : FAILURE in 4m 29s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5662a0ebdbe74a0ab0c96fb823eb1ab7 : SUCCESS in 23m 52s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/ec8dc490b53946de9efa02ffb8edd80b : SUCCESS in 20m 11s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/91f220cbd1b54f3d9b3361c4e63281f9 : SUCCESS in 24m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b1054a236be4f791fd210948c20fb7042b782b06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 03:24:13.000000000","message":"Topic set to migrate-to-noble","accounts_in_message":[],"_revision_number":1},{"id":"959be4e8bd5e762cfe99ba5e1f02209f42404ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-12 03:23:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8e6a0cec3c20fa268748a264bf2e846ef6233f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 03:24:01.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"4677d8d7a2afbbb37e85131178d25cf03253b628","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-14 05:58:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82dd2a8d974e3b57e5d4eda601ce5f0328359d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 05:58:28.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"1904c32a535c1267df9c466d6f567731338df48e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-15 01:22:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ddcc813cd45c77fdd52d8c9b5c79819e19f43b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 01:22:26.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"a1354dddc162e83a58331ae486bba2780cc83e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-15 01:39:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"29f56531ace5589188bdb40d83648d175bae79f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 02:26:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3565cca0c440b9b7bf90162f9d7685\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c672367ed753440aa31f3a9f985296e8 : SUCCESS in 2m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1cd5ab2d7ebb4f5787263088a2bd1160 : SUCCESS in 2m 34s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/352510afdc354d47bea3257acac89e90 : SUCCESS in 38m 24s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/eac95a2997404d0aa5b695104617d9f1 : SUCCESS in 18m 29s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9069e49b3cfd44fbb3639a1f66f27c65 : SUCCESS in 36m 34s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/c8b20c5467ba46bfbc9e0efe441c04e1 : SUCCESS in 19m 34s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/fbe706c8258c4102815832e6f5ea5bdb : SUCCESS in 23m 35s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/f9046b367746423e85dfdf9253e40731 : SUCCESS in 23m 42s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/197b572ba0e847f2a127c7f17cef86bf : SUCCESS in 38m 49s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/c1f70c8b0af04984b7ba27c55065d9a9 : SUCCESS in 40m 55s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/47c314683d9a47bca77a68913c86c8fa : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":3},{"id":"2790eb3ec7c1a7ee7132b0155670d70fa8997c4c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-11-15 10:25:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"266676daed1553e66a00228a04d396f48794c1ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-17 02:54:57.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2aa9875df20e9489e5f18d53c30d58fee0992f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-17 03:40:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a69d1a05384828990787eba7668212\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ab3ac92a4443608c2498fd0dc1112d : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/465771bf6af44736922c1791e5871363 : SUCCESS in 1m 43s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f4ff75dae73e4d89a8837f775169a998 : SUCCESS in 29m 02s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/6195a73b114547e28e68bd01acdc08a6 : SUCCESS in 29m 42s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/99c5a3478d6844a094fbdc2c9e2ba088 : SUCCESS in 28m 30s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/51a6f4d820f2457cabdea415d8a0365f : SUCCESS in 33m 15s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/289040ef850e406aa7b779cdb8979c23 : SUCCESS in 36m 39s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/4c2638947aae48c2a4f214968305b74f : SUCCESS in 26m 14s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d627e8ac3d2248368553e4a41e1b7893 : SUCCESS in 25m 33s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/21f30368de79413ea03a70cce1a6d615 : SUCCESS in 28m 20s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/280be02e53054fa88a553d5ea1d29daf : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":4},{"id":"9fa642a94ef605fac13f8821ff1a69f4601c0853","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-18 16:19:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53f6ff140a019af74acf424c37722a0cb7a91422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 16:19:26.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f45a68a215e749b2932736ee8f51ca18\n\n\nWarning:\n  Change 932156 in project openstack/tempest does not share a change queue with 932946 in project openstack/barbican-tempest-plugin","accounts_in_message":[],"_revision_number":4},{"id":"97e116cdbda2d62685faceb0b3dae9725e3174f8","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-11-18 16:58:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c0091683cf8709af41ddd64b5cadce1c7cd71f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 17:42:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b863114dc8d8478bb7d39f3ace5ad8cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b429e5781f49488d14d0b9f90f2ddc : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02453d5031fe472a92e871d4a988e598 : SUCCESS in 1m 35s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/668ee636283a4af4a40e666e75d51369 : SUCCESS in 31m 05s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/68de44a022e1435cbd34c4d1f40c6a31 : SUCCESS in 20m 07s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/aca41a8675ce4ad9bb0befb37c27b466 : FAILURE in 25m 49s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/cfea46065b974447977ad59f95c3e68d : SUCCESS in 27m 29s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/b726043f6e3448409ae7d3eb8422b9a6 : SUCCESS in 24m 58s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/698afb5f94af4c19ae3287a4867f87b0 : SUCCESS in 24m 58s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/05e4f775263b476f816331752768b4d3 : SUCCESS in 37m 04s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/f5ef75ffb6184c26a0aa5d1b565927e6 : SUCCESS in 33m 37s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/5119d177179749159fdeea6c58000555 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":4},{"id":"cf2dfc7636ce3f0a84c15f6c1778f38548a77022","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-30 18:45:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a46e9797bff39084cf769a87d08ca6d370b847a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 19:28:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1999acde8494f8da6677f121d8a9dae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1014c23863824cb1a128599ad6c65813 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f02b5a2c475485a8ff693a0488da7a8 : SUCCESS in 2m 15s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/146e5753144c423b9404aa289af1b503 : SUCCESS in 23m 16s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/1936b562f36e4117be60eb16371f8cb4 : SUCCESS in 33m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/e2ae301888734fba990cebe544043f39 : SUCCESS in 25m 46s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/217b940b632647f0a1f418b2eeba848e : SUCCESS in 36m 24s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/3b39634dc50a4765afeb942beeb1370b : SUCCESS in 24m 47s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/555ab197e7a845bdbb04a32e7728a462 : SUCCESS in 23m 47s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d831f6ee472b44e793ecd52a2cab93c3 : SUCCESS in 36m 43s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/b55c321435c14c38b33413604992ed3d : SUCCESS in 23m 23s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/b69a382fd8a84cd288e1fd9f4c0c3c89 : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fd8492f542e9a395ed9aebb04dd4b78302370a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 19:29:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"033197196821f07fbdae51bfb5a88919e94a34ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 20:11:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eacb4ae053084a36bde9c05e4ea3b986\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8baa5e96bfdf484fa925d60e5f093322 : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c93ed707e52348beb9512c795e0c44c7 : SUCCESS in 3m 08s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6ff586bd330a466eb40ff1edac7f9e51 : SUCCESS in 36m 26s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/1b06f0be1cc24973ba0533ea2e7fec45 : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":4},{"id":"0a82515b10c6181d8ba5ac4d62f55c5c6efc800d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 20:11:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c6da0009810199f755c2bc69bffcd615975e118e","revisions":{"b71d966861537c4f69e0ab952e56e88bcbad16df":{"kind":"REWORK","_number":1,"created":"2024-10-22 01:30:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/932946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/46/932946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/1"}}},"commit":{"parents":[{"commit":"d7f96744c352e16301832751ae4af14818c4ec5c","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/d7f96744c352e16301832751ae4af14818c4ec5c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 01:27:58.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 01:30:34.000000000","tz":-420},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I86b6abfdbd06851f6d1b331ce113e0897384bd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b71d966861537c4f69e0ab952e56e88bcbad16df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/b71d966861537c4f69e0ab952e56e88bcbad16df"}]},"branch":"refs/heads/master"},"3eb4487ade07aa87a95746ebd4f6688a873001d1":{"kind":"REWORK","_number":2,"created":"2024-11-12 03:23:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/932946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/46/932946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/2"}}},"commit":{"parents":[{"commit":"ecd7e8abbd60672eaff290cac896f6a98543ad52","subject":"Remove stable/2023.1 job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ecd7e8abbd60672eaff290cac896f6a98543ad52"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 03:23:05.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 03:23:05.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I86b6abfdbd06851f6d1b331ce113e0897384bd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3eb4487ade07aa87a95746ebd4f6688a873001d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/3eb4487ade07aa87a95746ebd4f6688a873001d1"}]},"branch":"refs/heads/master"},"f23302d0627579cc0ff32d4a55ee6765294c3fd0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-15 01:39:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/932946/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/46/932946/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/3"}}},"commit":{"parents":[{"commit":"c01cab3614ee2924babde7cf36c13e1029a1188e","subject":"Remove stable/2023.1 job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c01cab3614ee2924babde7cf36c13e1029a1188e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 03:23:05.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-15 01:39:40.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I86b6abfdbd06851f6d1b331ce113e0897384bd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f23302d0627579cc0ff32d4a55ee6765294c3fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/f23302d0627579cc0ff32d4a55ee6765294c3fd0"}]},"branch":"refs/heads/master","description":"Rebase"},"c6da0009810199f755c2bc69bffcd615975e118e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-11-17 02:54:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/932946/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/46/932946/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/46/932946/4"}}},"commit":{"parents":[{"commit":"c01cab3614ee2924babde7cf36c13e1029a1188e","subject":"Remove stable/2023.1 job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c01cab3614ee2924babde7cf36c13e1029a1188e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 03:23:05.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-17 02:54:56.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nCloses-Bug: #2088356\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I86b6abfdbd06851f6d1b331ce113e0897384bd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c6da0009810199f755c2bc69bffcd615975e118e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c6da0009810199f755c2bc69bffcd615975e118e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
