)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~935377","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I77b4758fc025f076d40146140008dc4aff2c0527","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-11-15 16:58:01.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I77b4758fc025f076d40146140008dc4aff2c0527","subject":"Add job for testing PKCS#11 backend with SoftHSM","status":"NEW","created":"2024-11-15 16:57:21.000000000","updated":"2024-12-02 16:04:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26867c0cc33b6bae2c0918b146c5af01a34fc1a1","_number":935377,"virtual_id_number":935377,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2024-12-02 16:04:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2024-12-02 15:34:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2024-12-02 15:33:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2024-12-02 15:34:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 16:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 15:33:11.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2024-12-02 15:34:07.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"432c85224f091ceb8238cf39293393cedd61f3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-15 16:57:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54393933f097a5c9444c59c317b7737c92c3428e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:58:01.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/barbican-tempest-plugin on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027devstack_localrc\u0027]\n\nThe problem appears in the the \"barbican-tempest-plugin-pkcs11-softhsm\" job stanza:\n\n  job:\n      name: barbican-tempest-plugin-pkcs11-softhsm\n      parent: barbican-tempest-plugin-simple-crypto\n      devstack_localrc:\n        ENFORCE_SCOPE: true\n      devstack_services:\n        barbican-softhsm: true\n\n  in \"openstack/barbican-tempest-plugin/.zuul.yaml@master\", line 61","accounts_in_message":[],"_revision_number":1},{"id":"b3623de720c0d9e5a3f23fd3b9f83292fd9fddcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-15 16:59:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d00620be27977aaf0778875eb747e62f9ceaa480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:01:15.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/barbican-tempest-plugin on branch master.  The\nproblem was:\n\n  Job barbican-tempest-pluign-pkcs11-softhsm not defined\n\nThe problem appears in the the \"opendev.org/openstack/barbican-tempest-plugin\" project stanza:\n\n  project:\n      queue: barbican\n      templates:\n        - tempest-plugin-jobs\n        - check-requirements\n      check:\n        jobs:\n          - barbican-tempest-plugin-simple-crypto\n          - barbican-tempest-plugin-simple-crypto-secure-rbac\n  ...\n\n  in \"openstack/barbican-tempest-plugin/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":2},{"id":"1f9c642ffb6ef9ca134c82718bf6d66ed910aeb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-15 17:02:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bfc8c285f0f40b01d5843cee5f2e0d783637bb2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:45:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d389db855dde455b8ec527b8d1db1dbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb5909a2c2e48d08281e6149c7fd5a4 : SUCCESS in 2m 25s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0748e0af2181473b8a716b42525fd4a1 : SUCCESS in 16m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/ecfe638b36e44b96819cd82c14314d92 : SUCCESS in 23m 47s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/f9be3408a3a24586a4b00b5ef95aa92e : SUCCESS in 16m 52s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/72863eb624c04902912016ae61a964c5 : SUCCESS in 25m 05s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/456f921e6c8945a9b06d57285b94b15b : SUCCESS in 34m 35s\n- barbican-tempest-plugin-simple-crypto-2023-1 https://zuul.opendev.org/t/openstack/build/f6c299b69d7b4db4b680526171083e1a : SUCCESS in 19m 27s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/717976e43f8d4ad7af68e49a21a07dd6 : SUCCESS in 29m 41s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/dc523014aabf4dc88294f184b97ad57d : SUCCESS in 23m 39s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/1cf3215ac4d74e78bc899c70511f2adb : SUCCESS in 35m 39s\n- barbican-tempest-plugin-pkcs11-softhsm https://zuul.opendev.org/t/openstack/build/878adc8d124e4fc1be59a7bb5a4ee736 : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b5d68de94d5a8964a69b670ff2d99e6ad8dfdc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-18 16:19:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ba105435ac6babfef43e69438872953f5e77adfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 17:01:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/998bfa0ad4064d12864af7938dc63167\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6b1f98e19d4cd79e7fea99e8de5077 : SUCCESS in 2m 42s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f881c4c7eb9846099c77ca38af13a3de : FAILURE in 7m 42s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/4e9f3cce1e654b58ac3c196033a3b1d4 : SUCCESS in 36m 46s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/442151745ef0464d94617721f30983d7 : FAILURE in 7m 54s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/8a63c1d52b844125ab4d030ba512e6ab : SUCCESS in 21m 23s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/37c6df220e6f48e485997fd6c46dec36 : SUCCESS in 23m 11s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/a09d107da7704972b3c26febf2a6547d : SUCCESS in 39m 31s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6e5d8e0a6cff40738d22a65d2514f49e : FAILURE in 18m 13s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/66e9e73000c74596a1494fb59e08979a : FAILURE in 6m 46s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/a865481a2ee74c7e80226c470d71d911 : FAILURE in 18m 47s\n- barbican-tempest-plugin-pkcs11-softhsm https://zuul.opendev.org/t/openstack/build/a29f75badf414ce8b72220d9d1370731 : FAILURE in 18m 57s","accounts_in_message":[],"_revision_number":4},{"id":"9b49a50384a6f363a4aaa7ceb439e8ed40ea095e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-22 15:19:06.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"74b2a54f0ede5058dfbcc4367b0b225a2bf8620f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 16:58:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb4dfd8c7164dbeb4be2650d86209e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b85e39a32a4a4b9121cd788aa8daea : SUCCESS in 2m 42s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f7ab745704a245b1bf8021e06ec40f49 : SUCCESS in 32m 23s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c0c517ceba2846f2bc6b6b8552f08450 : SUCCESS in 35m 30s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/b9a9f8868d5a4c6a9d4a37c6f663f79e : SUCCESS in 31m 10s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/f35670471fd94828af3e8d55be355eb1 : SUCCESS in 35m 41s\n- barbican-tempest-plugin-simple-crypto-2023-2 https://zuul.opendev.org/t/openstack/build/d962022cedc847b98ce3b7fc66a6f8ae : SUCCESS in 40m 29s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9ef003e768a248dd96db23ac9400c249 : SUCCESS in 31m 53s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/c6fa8b7595bd40d683b1c4d5ef82b341 : SUCCESS in 29m 37s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/88ea188ad83347b7bb7b5578a4498a2d : SUCCESS in 33m 18s\n- barbican-tempest-plugin-pkcs11-softhsm https://zuul.opendev.org/t/openstack/build/b0a756ee46584247ac4d9e6c43a2630e : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":5},{"id":"26f3b0b5baa6efd1b1693590acbc1e7042d99dd5","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2024-12-02 15:33:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"db209ab05180f81f5430756495d2a1224dfe3a2f","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-02 15:34:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8e73dfacc70dd09c252251f1b6a01f9fe33a8fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 15:35:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"26867c0cc33b6bae2c0918b146c5af01a34fc1a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 16:04:00.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/935208 is needed.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c7b43585510f3f0dc077d5dc6348d07726d795df","revisions":{"5c9a55bac5ab0df4d488003b0eead89a0554439c":{"kind":"REWORK","_number":1,"created":"2024-11-15 16:57:21.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/935377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/77/935377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/1"}}},"commit":{"parents":[{"commit":"df4cb84e52cc78b45489d1a08773e80080a9621e","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/df4cb84e52cc78b45489d1a08773e80080a9621e"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:56:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:57:16.000000000","tz":-300},"subject":"Add job for testing PKCS#11 backend with SoftHSM","message":"Add job for testing PKCS#11 backend with SoftHSM\n\nThis patch adds a new gate job to test the Barbican PKCS#11 backend with\nSoftHSM.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/935208\nChange-Id: I77b4758fc025f076d40146140008dc4aff2c0527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5c9a55bac5ab0df4d488003b0eead89a0554439c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/5c9a55bac5ab0df4d488003b0eead89a0554439c"}]},"branch":"refs/heads/master"},"7349032762366b70d8c2f072efb28f668df88f25":{"kind":"REWORK","_number":2,"created":"2024-11-15 16:59:51.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/935377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/77/935377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/2"}}},"commit":{"parents":[{"commit":"df4cb84e52cc78b45489d1a08773e80080a9621e","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/df4cb84e52cc78b45489d1a08773e80080a9621e"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:56:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:59:42.000000000","tz":-300},"subject":"Add job for testing PKCS#11 backend with SoftHSM","message":"Add job for testing PKCS#11 backend with SoftHSM\n\nThis patch adds a new gate job to test the Barbican PKCS#11 backend with\nSoftHSM.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/935208\nChange-Id: I77b4758fc025f076d40146140008dc4aff2c0527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7349032762366b70d8c2f072efb28f668df88f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/7349032762366b70d8c2f072efb28f668df88f25"}]},"branch":"refs/heads/master"},"369cd5e05a981c8e70e701f84b8131eee6a1101a":{"kind":"REWORK","_number":3,"created":"2024-11-15 17:02:04.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/935377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/77/935377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/3"}}},"commit":{"parents":[{"commit":"df4cb84e52cc78b45489d1a08773e80080a9621e","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/df4cb84e52cc78b45489d1a08773e80080a9621e"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:56:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 17:01:55.000000000","tz":-300},"subject":"Add job for testing PKCS#11 backend with SoftHSM","message":"Add job for testing PKCS#11 backend with SoftHSM\n\nThis patch adds a new gate job to test the Barbican PKCS#11 backend with\nSoftHSM.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/935208\nChange-Id: I77b4758fc025f076d40146140008dc4aff2c0527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/369cd5e05a981c8e70e701f84b8131eee6a1101a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/369cd5e05a981c8e70e701f84b8131eee6a1101a"}]},"branch":"refs/heads/master"},"520d0743dd37672fe8e4923ff3303406165789af":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-18 16:19:29.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/935377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/77/935377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/4"}}},"commit":{"parents":[{"commit":"c6da0009810199f755c2bc69bffcd615975e118e","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c6da0009810199f755c2bc69bffcd615975e118e"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:56:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-11-18 16:19:29.000000000","tz":0},"subject":"Add job for testing PKCS#11 backend with SoftHSM","message":"Add job for testing PKCS#11 backend with SoftHSM\n\nThis patch adds a new gate job to test the Barbican PKCS#11 backend with\nSoftHSM.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/935208\nChange-Id: I77b4758fc025f076d40146140008dc4aff2c0527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/520d0743dd37672fe8e4923ff3303406165789af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/520d0743dd37672fe8e4923ff3303406165789af"}]},"branch":"refs/heads/master","description":"Rebase"},"c7b43585510f3f0dc077d5dc6348d07726d795df":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-22 15:19:06.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/935377/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/77/935377/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/77/935377/5"}}},"commit":{"parents":[{"commit":"c01cab3614ee2924babde7cf36c13e1029a1188e","subject":"Remove stable/2023.1 job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c01cab3614ee2924babde7cf36c13e1029a1188e"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-15 16:56:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-11-22 15:19:06.000000000","tz":0},"subject":"Add job for testing PKCS#11 backend with SoftHSM","message":"Add job for testing PKCS#11 backend with SoftHSM\n\nThis patch adds a new gate job to test the Barbican PKCS#11 backend with\nSoftHSM.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/935208\nChange-Id: I77b4758fc025f076d40146140008dc4aff2c0527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c7b43585510f3f0dc077d5dc6348d07726d795df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/c7b43585510f3f0dc077d5dc6348d07726d795df"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
