)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~951231","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I86ea4e47d43daad02f285d4fe85dd3c76f539f6c","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-17 16:18:20.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-09-17 15:01:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I86ea4e47d43daad02f285d4fe85dd3c76f539f6c","subject":"Remove stable/2023.2 job","status":"MERGED","created":"2025-05-29 05:43:19.000000000","updated":"2025-09-17 16:18:20.000000000","submitted":"2025-09-17 16:18:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951231","meta_rev_id":"de3c4e1431bcaf579991a2c7a734b4a446659730","_number":951231,"virtual_id_number":951231,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-17 16:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"recommended":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2025-07-25 14:06:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2025-09-17 15:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":1,"date":"2025-09-17 15:01:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 05:45:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-05-29 07:09:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-25 14:06:06.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"cdeaba87f83078a5067b611ad91643d05497b4ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 05:43:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dbeceafc5cf61b7138a1735dc669245968b9930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 07:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee98ea778b74d58816750ad7d235e45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9551b751296a46d2aef21ac37c0309e4 : SUCCESS in 3m 21s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2a36d955447b41cd8ceb54e8f60b76a0 : SUCCESS in 22m 40s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/712482cb1f944754918b1eb67f6c01ce : SUCCESS in 28m 52s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/0fb9bfdffc564480bc91f84a011ee859 : SUCCESS in 33m 58s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/6ee0e34c738a43ffa1e821cdbfe28f2c : SUCCESS in 26m 11s\n- barbican-tempest-plugin-simple-crypto-2024-1 https://zuul.opendev.org/t/openstack/build/1e3d9ef87b964406894d9c24015838d6 : SUCCESS in 28m 38s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/4432a1ac28c446de8638240d6f780a96 : SUCCESS in 31m 36s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/3193215413c64ecf848a7240ba0c39f4 : SUCCESS in 33m 04s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f5e57e1e6e9a4fce878e452b28054071 : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":1},{"id":"53d2325c9b78974e5ad6c61c9edc1ba72f06d2c0","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-07-25 14:06:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"06a46e4e2bca1a660f6bc76618ec74fa7d90a7f9","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-09-17 15:01:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63813a8cab0679aef8b36d9611afd0237c69fd36","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-09-17 15:01:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a975c6b8161c50df91122c9514db00f8808c9f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:02:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"565877318e5c941ff383b4d4a69c1285d3a024ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 16:18:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d2adccec77487881aaf11269695402\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c001d3eb2584c15977efae44b5831a8 : SUCCESS in 3m 48s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0548c3b54a440ff89a02fa0493f84d2 : SUCCESS in 30m 50s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/45de27c7716f4fb4b164ec484334abd8 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"de3c4e1431bcaf579991a2c7a734b4a446659730","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 16:18:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"debd1648449976dd92d9df165af19e11bd43424a","revisions":{"debd1648449976dd92d9df165af19e11bd43424a":{"kind":"REWORK","_number":1,"created":"2025-05-29 05:43:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/951231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/31/951231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/31/951231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/31/951231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/31/951231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/31/951231/1"}}},"commit":{"parents":[{"commit":"2cd185133a86242c58d721e66c2db753a369fe60","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/2cd185133a86242c58d721e66c2db753a369fe60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:40:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:40:41.000000000","tz":540},"subject":"Remove stable/2023.2 job","message":"Remove stable/2023.2 job\n\n... because stable/2023.2 reached its EOL.\n\nChange-Id: I86ea4e47d43daad02f285d4fe85dd3c76f539f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/debd1648449976dd92d9df165af19e11bd43424a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/debd1648449976dd92d9df165af19e11bd43424a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
