)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~966367","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Iceacbd7c67ccfb1fc8aaa233b3d99d4af91c9510","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"2024-1-last","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-01 18:20:58.000000000","reason":"Change was abandoned"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-12-01 18:20:58.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-12-01 18:20:55.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Iceacbd7c67ccfb1fc8aaa233b3d99d4af91c9510","subject":"Update stable branch jobs","status":"ABANDONED","created":"2025-11-06 22:15:46.000000000","updated":"2025-12-01 18:20:58.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52dfd8c0cfdb0d59f88c26e75c8f9278e76388cb","_number":966367,"virtual_id_number":966367,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-25 20:10:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2025-11-26 13:19:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 23:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 15:15:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-11-25 19:30:34.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-11-26 13:19:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"0794ae979408da48179bf45f5a9addab509711de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4427482ce56ba799fdbdb18d9db0556cda60712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:00:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e92608d7889245ea8d55e0418e769e35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd56d464dac42ed9dc9b98e40fe0511 : SUCCESS in 2m 03s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d37de5dba7c644cd86155b704e76d759 : SUCCESS in 34m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/3fb95812409443fb92b05a7831834079 : SUCCESS in 33m 13s\n- barbican-tempest-plugin-simple-crypto-2025-2 https://zuul.opendev.org/t/openstack/build/0069d7d1f9b94a269c138c693eb1ff13 : FAILURE in 22m 58s\n- barbican-tempest-plugin-simple-crypto-2025-1 https://zuul.opendev.org/t/openstack/build/09e67e9ee93f4a2887f1b326c4be8821 : FAILURE in 32m 32s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/b2f7c0d26ac841abb03b63d002ab6c3a : FAILURE in 26m 29s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd69026f1774de88b3d1c3d1b972d53 : SUCCESS in 18m 26s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/63679e313d784b9db92bba9ca6a81c94 : SUCCESS in 33m 07s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/57b3524af83847e798d2cff1bcc735ae : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cd5e3d468e096acc8191f4194039bbdc1e234a36","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:43:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02783544d58e80b12effbaa9abba9184a1d9e4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:26:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7e0b5dfd6df4139bc386ac1f01d202e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e5dcbf628242c1a2e307426475079d : SUCCESS in 2m 54s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a60fb35a13747368e1044d46b605082 : SUCCESS in 25m 47s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/10a39b4419e54456b89e2efda2bb8a4c : SUCCESS in 34m 20s\n- barbican-tempest-plugin-simple-crypto-2025-2 https://zuul.opendev.org/t/openstack/build/a92aa970bbd847dca4194bd4a8dcf2e0 : SUCCESS in 28m 15s\n- barbican-tempest-plugin-simple-crypto-2025-1 https://zuul.opendev.org/t/openstack/build/f1441b3d41694621abe95ecb83647f8a : SUCCESS in 31m 27s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/f5f60e038e404570839c019483e77fcd : SUCCESS in 23m 18s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/92fa4f6270c546dea46cb036b619c020 : SUCCESS in 29m 14s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/632e9fe801ca4782b2f28b034badcef4 : SUCCESS in 31m 09s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/8c3a8f14ac864fb58741e2995a77e17f : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":1},{"id":"31c98f53f99c9306cf92ffc217244c45c60b77b6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 15:15:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7ac0e785c50e5adda0251fc8a719b1f3ad20a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-25 19:30:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"97792dac0c11e5707b086eb99d4ed7f557d1bcb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 20:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff60378408c4a54a14cf5c80a66d508\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8cba8c50c8a048ce8a64f6b2e48bc719 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a0aabf22e0405f81809f82e42c3061 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9953673b871a4366a368b346d3b78ac3 : SUCCESS in 2m 10s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6002e6817c5b4907b0ae49041046e339 : SUCCESS in 17m 52s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/6b92616cd0d24e6fb74dc85e0065fc4f : SUCCESS in 15m 04s\n- barbican-tempest-plugin-simple-crypto-2025-2 https://zuul.opendev.org/t/openstack/build/f96406ad37a647bcb7b76e8397ad2389 : SUCCESS in 33m 56s\n- barbican-tempest-plugin-simple-crypto-2025-1 https://zuul.opendev.org/t/openstack/build/d23a271ac24f4f07bef9b1c89ebc70ee : SUCCESS in 33m 43s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/f8a29882422b4fc69a7a5cc107d02ed2 : SUCCESS in 26m 20s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c974092581d043ea9d641db914f672fe : SUCCESS in 17m 40s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/2372c32218d648cb895240f4a3f1adc0 : SUCCESS in 23m 28s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/f3777e1d246249109b40cc2ef0f03c5e : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f96eb1e8dd068d8d6c6b9435c19fe26d81faeb80","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-26 13:19:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"edd570ae6e5129637819586f8c964f010144cc94","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-12-01 18:20:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52dfd8c0cfdb0d59f88c26e75c8f9278e76388cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-12-01 18:20:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce2f85d8585edf9cd452546dd106af2fc80a24c9","revisions":{"716ff1e3e7dc52290c7f7b0f85c83b90fe21050e":{"kind":"REWORK","_number":1,"created":"2025-11-06 22:15:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/966367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/67/966367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/1"}}},"commit":{"parents":[{"commit":"6afff90c31e0a66a709f74b66c95a2391be6ca16","subject":"Remove Ubuntu 22.04 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/6afff90c31e0a66a709f74b66c95a2391be6ca16"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:42:34.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 22:09:51.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs.\n\nChange-Id: Iceacbd7c67ccfb1fc8aaa233b3d99d4af91c9510\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/716ff1e3e7dc52290c7f7b0f85c83b90fe21050e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/716ff1e3e7dc52290c7f7b0f85c83b90fe21050e"}]},"branch":"refs/heads/master"},"ce2f85d8585edf9cd452546dd106af2fc80a24c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-25 19:30:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/966367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/67/966367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/67/966367/2"}}},"commit":{"parents":[{"commit":"82a83c0c759e0cc168022bc03ca3461fe3417358","subject":"Remove Ubuntu 22.04 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82a83c0c759e0cc168022bc03ca3461fe3417358"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:42:34.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-11-25 19:30:25.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs.\n\nChange-Id: Iceacbd7c67ccfb1fc8aaa233b3d99d4af91c9510\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ce2f85d8585edf9cd452546dd106af2fc80a24c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/ce2f85d8585edf9cd452546dd106af2fc80a24c9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
