)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~980983","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~I96082b0a73e772fac1e98e4cc921a88cd0ff6dd5","project":"openstack/barbican-tempest-plugin","branch":"master","topic":"2026.1-job","attention_set":{},"removed_from_attention_set":{"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2026-04-20 08:57:48.000000000","reason":"\u003cGERRIT_ACCOUNT_35125\u003e replied on the change","reason_account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-20 09:19:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96082b0a73e772fac1e98e4cc921a88cd0ff6dd5","subject":"Add stable/2026.1 job","status":"MERGED","created":"2026-03-17 15:28:47.000000000","updated":"2026-04-20 09:19:43.000000000","submitted":"2026-04-20 09:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980983-2026.1-job","meta_rev_id":"d39f9f2ca6fb580fe14cde011155b25cba076671","_number":980983,"virtual_id_number":980983,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 09:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":2,"date":"2026-04-20 08:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2026-04-20 08:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 16:18:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 14:35:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"0d1e02c218e5c8d237767dc61d3c39c93fa87d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 15:28:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"358a14f255eb7ae36140ce0b156bbc70df78ac55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9816b1748f8d44f4966ffed019417c5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e280b283c144a68ea5faef16271997 : SUCCESS in 2m 46s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3465e0e2445b4b4aa92fcd2492fb4b7f : SUCCESS in 26m 01s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/dc04f64aec204d6483586bbac6a93897 : SUCCESS in 37m 59s\n- barbican-tempest-plugin-simple-crypto-2026-1 https://zuul.opendev.org/t/openstack/build/8bcd317dcb344a18866f3be32c040c13 : SUCCESS in 17m 54s\n- barbican-tempest-plugin-simple-crypto-2025-2 https://zuul.opendev.org/t/openstack/build/acc9d6cf01fb40e4ae72d38cd7779a10 : SUCCESS in 18m 22s\n- barbican-tempest-plugin-simple-crypto-2025-1 https://zuul.opendev.org/t/openstack/build/14e4f0130b9048278ce5e6ef3255079c : SUCCESS in 34m 49s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/6ef6d7a9f6d142f28a0140e97c0efff9 : SUCCESS in 16m 18s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6471cd3ba346493f84a8073c19dcc630 : SUCCESS in 34m 44s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/c8a1ff2a05ca4bc3a4c5b4b1bc59d316 : SUCCESS in 19m 03s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/31947a5cd7504460bbcd9b608ceda974 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"83d090a749b80265ef4f8323bafbfb3e0dbd0b2c","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-04-20 08:57:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37d06f13c3d35979a20b836d8c76ee5aa8f70629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 08:58:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fbbc8abbe756afd9070f9ab30cb546282c31275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:19:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2473169f992b40518336a885b368d6fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468f2b280c0f4a418abff819cdf92b90 : SUCCESS in 1m 59s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ca25659fcd0041bd91b461a0f32d2b70 : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d39f9f2ca6fb580fe14cde011155b25cba076671","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:19:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fbaa67384b61ee2932f86a53170cf162ffbae73","revisions":{"4fbaa67384b61ee2932f86a53170cf162ffbae73":{"kind":"REWORK","_number":1,"created":"2026-03-17 15:28:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/980983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/83/980983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/83/980983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/83/980983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/83/980983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/83/980983/1"}}},"commit":{"parents":[{"commit":"82a83c0c759e0cc168022bc03ca3461fe3417358","subject":"Remove Ubuntu 22.04 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82a83c0c759e0cc168022bc03ca3461fe3417358"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:28:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:28:30.000000000","tz":540},"subject":"Add stable/2026.1 job","message":"Add stable/2026.1 job\n\nThe branch was cut, so should be tested.\n\nAlso fix the wrong branch used for 2025.2 job.\n\nChange-Id: I96082b0a73e772fac1e98e4cc921a88cd0ff6dd5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4fbaa67384b61ee2932f86a53170cf162ffbae73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/4fbaa67384b61ee2932f86a53170cf162ffbae73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
