)]}'
{"id":"openstack%2Fbarbican-tempest-plugin~986040","triplet_id":"openstack%2Fbarbican-tempest-plugin~master~Ia00d6af1dbef8bdcc84a8bb73dcc7574530c0daf","project":"openstack/barbican-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-15 13:23:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia00d6af1dbef8bdcc84a8bb73dcc7574530c0daf","subject":"Drop direct reference to [identity] auth_version option","status":"MERGED","created":"2026-04-24 07:22:26.000000000","updated":"2026-05-15 13:23:28.000000000","submitted":"2026-05-15 13:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986040","meta_rev_id":"6c89aedc6ef82499d34fe76895b2426bf7034fc4","_number":986040,"virtual_id_number":986040,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 13:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":2,"date":"2026-05-15 12:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2026-05-15 12:43:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 08:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 12:43:06.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"8aa0819b9ad071cb32d18d3575b2ebbe96587343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-24 07:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"419fa08c6fbff7db066b4f83b49ad20dc9b8c38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-24 07:24:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4c5aa1112bdbd52c0ddecc14d601e68565544391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 08:10:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056065d9e0704036aa49621297e2c623\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1aea756fb1046c88bac664f0ae6d355 : SUCCESS in 2m 05s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/06e2b19e9be74154a3855a95625e22a3 : SUCCESS in 19m 45s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/9093b5dcf8ee41cc80a5f031020851ab : SUCCESS in 19m 30s\n- barbican-tempest-plugin-simple-crypto-2026-1 https://zuul.opendev.org/t/openstack/build/0e27a0ddeafb4dc08543a11de37266d1 : SUCCESS in 19m 17s\n- barbican-tempest-plugin-simple-crypto-2025-2 https://zuul.opendev.org/t/openstack/build/023144ae42d24252985781de389b2f6a : SUCCESS in 39m 40s\n- barbican-tempest-plugin-simple-crypto-2025-1 https://zuul.opendev.org/t/openstack/build/df0137aa88694483af6e3022264a3c64 : SUCCESS in 34m 06s\n- barbican-tempest-plugin-simple-crypto-2024-2 https://zuul.opendev.org/t/openstack/build/fd24753871ca4346bf5e692d0d48d0f7 : SUCCESS in 30m 35s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/256a4709c59e48a593922d55f79f04fb : SUCCESS in 34m 49s\n- barbican-tempest-plugin-simple-crypto-castellan-src https://zuul.opendev.org/t/openstack/build/9671fc79cca14943bb9ad8c59df041c8 : SUCCESS in 34m 15s\n- barbican-tempest-plugin-simple-crypto-cursive https://zuul.opendev.org/t/openstack/build/646f5293f5154a46b2c04a3950cdf6e8 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f5228e694e128152dd97b5108cff9cb192b20afe","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-15 12:43:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nNecessary for forward-compatibility with Tempest removing the auth_version option. Clean fix.","accounts_in_message":[],"_revision_number":2},{"id":"84ccb469af5d61d0ebde86bbf1c70975df6984ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:43:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc3b7022b4ab4c334c399dccac4a428e774f75e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:23:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c968c88a520c481e85a5eb4ba7a1e18a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86059ef2e8c34b3eac151d12bd1d5c34 : SUCCESS in 4m 15s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9a85e1c039a245a3a4f247459ffb4ccb : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6c89aedc6ef82499d34fe76895b2426bf7034fc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:23:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"184179653cf151204fb37b1b5b5008d63ac46fa5","revisions":{"57cd18f0c9985e076610e939edc32bd4ebfd0e70":{"kind":"REWORK","_number":1,"created":"2026-04-24 07:22:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/986040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/40/986040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/1"}}},"commit":{"parents":[{"commit":"82a83c0c759e0cc168022bc03ca3461fe3417358","subject":"Remove Ubuntu 22.04 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82a83c0c759e0cc168022bc03ca3461fe3417358"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:22:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:22:04.000000000","tz":540},"subject":"Drop direct reference to [identity] uri option","message":"Drop direct reference to [identity] uri option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: Ia00d6af1dbef8bdcc84a8bb73dcc7574530c0daf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/57cd18f0c9985e076610e939edc32bd4ebfd0e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/57cd18f0c9985e076610e939edc32bd4ebfd0e70"}]},"branch":"refs/heads/master"},"184179653cf151204fb37b1b5b5008d63ac46fa5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-24 07:24:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/986040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican-tempest-plugin","ref":"refs/changes/40/986040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican-tempest-plugin refs/changes/40/986040/2"}}},"commit":{"parents":[{"commit":"82a83c0c759e0cc168022bc03ca3461fe3417358","subject":"Remove Ubuntu 22.04 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/82a83c0c759e0cc168022bc03ca3461fe3417358"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:22:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:24:08.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: Ia00d6af1dbef8bdcc84a8bb73dcc7574530c0daf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/184179653cf151204fb37b1b5b5008d63ac46fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican-tempest-plugin/commit/184179653cf151204fb37b1b5b5008d63ac46fa5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
