)]}'
{"id":"openstack%2Fbarbican~680554","triplet_id":"openstack%2Fbarbican~master~Ied6cbc80c4b384ed45c927f2aea8e88604f457fe","project":"openstack/barbican","branch":"master","topic":"fix-docs-job","hashtags":[],"change_id":"Ied6cbc80c4b384ed45c927f2aea8e88604f457fe","subject":"Ensure doc/source/_static to fix docs gate","status":"MERGED","created":"2019-09-06 01:07:50.000000000","updated":"2019-09-06 14:36:41.000000000","submitted":"2019-09-06 14:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680554-1567780530135-1d1e633c","meta_rev_id":"1d4c1ffb8fcba7dc0e648415426080dff68c1984","_number":680554,"virtual_id_number":680554,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2019-09-06 01:11:53.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-09-06 14:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2019-09-06 13:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 09:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2019-09-06 13:22:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 09:08:37.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-09-06 13:22:39.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-09-06 14:35:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2771efc53e8a00e7b8c55a94765e04696032899","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 01:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05d20e0d074c673ee526eefeb6229b34b5466ece","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 01:11:53.000000000","message":"Patch Set 1:\n\nFYI: Recently cinder hit the same problem and I fixed it in https://review.opendev.org/#/c/678130/.","accounts_in_message":[],"_revision_number":1},{"id":"79293712b836a4863cc2e52b1f819357935b1ca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 02:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9eb9dc2a00948b6ba4a9673a4001d3a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4668ef7d818640b0941be11da585795f : SUCCESS in 3m 29s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/4281312021e940b48e2342c0a7973fd5 : SUCCESS in 50m 45s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/acd185d1a77140cab21ea195686fd03f : SUCCESS in 26m 49s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/0d1152aab2ae4ebeac3cf9035b5815ad : FAILURE in 31m 33s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/2a23d1853e9541a79e46fa75e5786993 : SUCCESS in 43m 48s\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/19bcccd6431545db8ce5cc11ecf70a47 : SUCCESS in 31m 08s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/7dab6410ca0c49bda73a1ec7ac8e51ae : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c008f54cf879186a419cc1af10b9e69f91f183d5","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-06 09:08:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9183741da6010c8bc9b9be4da4d06d1ef303bd3d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-09-06 13:22:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c4811248dcd509025f66b404060bfca56f30f02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:22:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03e273a1be482141ea380e687f2197b768f19ae6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:35:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c295abf3ea2e4a8699c65a18c3b338bd : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa39096eb6964a33a54f8d823ebe2eed : SUCCESS in 4m 02s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/d8ff510e3806453ab73824a166770c09 : SUCCESS in 43m 17s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/2efc5265433c4bf98fec8c14662a2ade : SUCCESS in 23m 48s\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/ecd97198a64c4cb49c7d2267af664f3f : SUCCESS in 35m 12s\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/2f31184623094d30b4c0c80386e15ca0 : SUCCESS in 26m 37s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/3d488d3ccd824b6985239ac8782b0ce2 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8a077873250ba7f5596e5e9679b86cd3fe05adbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:35:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e99bec292217e2e6f0e073ced74ee0993f5d399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:36:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d368922b3de450c9055773bd0a02fd1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d2376397d01b26ac2d98c0e02b67dfa0ecc2b1c","revisions":{"2d2376397d01b26ac2d98c0e02b67dfa0ecc2b1c":{"kind":"REWORK","_number":1,"created":"2019-09-06 01:07:50.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/54/680554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/54/680554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/54/680554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/54/680554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/54/680554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/54/680554/1"}}},"commit":{"parents":[{"commit":"ad7b3e6a219f7a33c3e26b222505ed4f1ccb1016","subject":"Merge \"Fix py3 compatibility issue in PKCS#11 plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ad7b3e6a219f7a33c3e26b222505ed4f1ccb1016"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-06 01:01:27.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-06 01:07:26.000000000","tz":540},"subject":"Ensure doc/source/_static to fix docs gate","message":"Ensure doc/source/_static to fix docs gate\n\ndoc/source/_static directory is required by oslo.policy sample generator.\nThe recent sphinx complains if _static directory is missing.\nThis commit adds a placeholder file to ensure doc/source/_static directory.\n\nChange-Id: Ied6cbc80c4b384ed45c927f2aea8e88604f457fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2d2376397d01b26ac2d98c0e02b67dfa0ecc2b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2d2376397d01b26ac2d98c0e02b67dfa0ecc2b1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
