)]}'
{"id":"openstack%2Fbarbican~694482","triplet_id":"openstack%2Fbarbican~master~I23b0859f81ab32158ef4364228c2bc782d7f6d2e","project":"openstack/barbican","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I23b0859f81ab32158ef4364228c2bc782d7f6d2e","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-11-15 09:01:27.000000000","updated":"2020-01-10 16:42:57.000000000","submitted":"2020-01-10 16:40:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694482-1578674434689-bbef5f2b","meta_rev_id":"0df2c3a5753d2a0325c6a92758e86ed61946cd4a","_number":694482,"virtual_id_number":694482,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 16:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2020-01-10 15:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2020-01-10 15:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":1,"date":"2020-01-04 22:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2020-01-10 15:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-04 22:13:25.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-01-10 15:21:32.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-01-10 15:29:02.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-01-10 16:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ae07bcd8c7fbf64c7398105c3c09b16dca4e6a6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-15 09:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"609678a505f10c9a536f050e2d2453df5e5c9df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:04:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c00cc3734e7409b87ab0294e38c9b43 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/205a150545c84903a3ccceb0df5c97db : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd8508f695f440e9d8bb250104e5725 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c84774f832b840e59bf1244ba0b2d84e : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e729ff42994728868bd0d9c9cb02ea : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/890977b4784a4d2ba49fe012ee371b26 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1068757ccb94109b913da436784bcaf : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16161820ae23440a80a417a7c842c306 : SUCCESS in 6m 43s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/52524877a1394cbc9772ec2b9cd11eba : FAILURE in 42m 01s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/bba526f8fcaf44988625dd9df14846ba : SUCCESS in 23m 32s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/7680701aade045a081974712d1ff9a43 : SUCCESS in 24m 40s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/a33da2f7ab4547f998622d6d22903bd7 : FAILURE in 35m 14s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/ea78328b585446f5809335646c596968 : SUCCESS in 24m 03s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/79346b29aa1845798bebcec280a2f56e : SUCCESS in 1h 05m 14s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03e975835f2848d4ad786d7b45c4f33a : SUCCESS in 51m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/cc52cc6fbd9a49c88ac20fb3b980f9c4 : SUCCESS in 1h 45m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4221879e78774e838bc9ae1b2e365f89 : SUCCESS in 1h 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caedd7c298ba46ea5ec02f1e607f18fb2d197ed5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-03 13:06:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4dde50e1a0f105e3013c024dd2b5124d308fd5dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 14:58:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55120524f71b44eb8487892b2bf365f4 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b3107e458d54673938dfcae425af74c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b692861486db4cd9b719dc19e1702e0b : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2acfbb1651984b0a9735d6bb3ec9e409 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72f0d5434ae486d98ae441ffc561224 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e35ba8a7c4174b0795aaefe73b766ed5 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d2dc8b03d64ae29d54ee27676e2022 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52bacfc593f141ddbb1e3adb7695500b : SUCCESS in 5m 54s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/1a0f5d8a23ae4674bc9581462071220b : SUCCESS in 38m 12s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/f357f0aa70bb4710a784c2bbb6bcd95d : SUCCESS in 20m 15s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/5df98a7fdbdb43dd94c0be5326b76820 : SUCCESS in 21m 16s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/55bda5958b5f4d8fbcada40339d6ffb0 : FAILURE in 31m 12s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/6ea5b903f32e468f8b5ba8d473571749 : SUCCESS in 21m 01s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/54bd3be329c444128dcae7147d6ff5c4 : FAILURE in 50m 54s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c253d9f6916e438a94274bb14affd128 : SUCCESS in 38m 48s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8a0dfe0fffd941e1a27d667b6f4a377a : SUCCESS in 1h 42m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69429c929a9347b3b3b59a0915e9846b : SUCCESS in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3e47c536ace4531b376af6216f7c5a51baf7e2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-03 21:51:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f084db824cb6fd75838e680a2f193103bbf433cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 23:50:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab4175ee4a03462ab9066fae576c5381 : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc5304514ffe4547bcdc3211d493eb41 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3eb5a2a0c847a4bbb926e9dc6d86e7 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cec8ea4b07d43b491e04835dff4db48 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d65f818b9649ee92b8b723b0cd4fea : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59634c6a027a4dd39bf1043594414965 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fce47fb49894bf3afb1da118fa8e177 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/687db80af4724e43a4ac7390b9e0e57a : SUCCESS in 5m 38s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/25102d3b74cb4184b889bf1b062e01d3 : SUCCESS in 38m 54s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/128b20275f504fbd87dcd40082274967 : SUCCESS in 27m 51s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/108cc176644b486ea0edf1ea6fc6bccf : SUCCESS in 22m 42s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/f95332b91a3b4880bc5133d3b7f28a66 : FAILURE in 39m 49s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/8ea223e0aeb7409e8eb4ca4e482b5ab4 : SUCCESS in 23m 56s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/71c96a9abe0a4cb293b75fd5bc86a764 : SUCCESS in 1h 05m 55s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3c13dd770d146d984b1e4d90c5cbc1f : SUCCESS in 38m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5f20acf89076436793f1f04e6f7d57b8 : SUCCESS in 1h 53m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3762518a0a1f41afae47ad6a789b5a01 : SUCCESS in 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5617950fbd6d00418bf4d19211f1f57050453a78","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-04 07:54:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b11ed8abf08b042f5eb2bdd9bb08411525b46b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 09:32:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8624ce32e6204c5f92ac3311c79ac77a : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e07c826a41f14af5be769082874a6204 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdcd1c19d56841919cec51695c5e2a5a : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/761f4b18a3834806bfc93aa1601dc70a : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d90263963f847f5b7133f05590e93fc : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1da531c4ee734e2687ae86d81b59e7f9 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b20fec411784d188a130133e0b3407d : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/524cf4f33f8747f99e398a94a1c9f61a : SUCCESS in 7m 07s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/e2f90698fa7942a2821a586b19ecdfa0 : SUCCESS in 36m 02s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/7a58fb290689437f8dac2fdb7ed0caf4 : SUCCESS in 23m 04s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/b4b67e5094c34150aa7f7b0f1fa7d581 : SUCCESS in 21m 02s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/7f403f4d3b11416a98f8e287445f5d24 : FAILURE in 40m 54s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/f80da70c9fb74960bd001bc382dbcd07 : SUCCESS in 20m 51s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/2d614a5f4adf4df09dbc8bf616b92dbd : FAILURE in 1h 13m 28s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3e731c17d0492ea70607dd65d1d7b8 : SUCCESS in 38m 59s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/05330e474b8d461a905a3a4f4dd56851 : SUCCESS in 1h 36m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc105cbf55d44a44b78af1e853d6016f : SUCCESS in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7537f6bc9fe9e6a824aa0f579b28ffc886834a6a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-04 09:48:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc1fe94e2f2afdec669370f9342510051a4c8431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 11:19:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e1b689a70249cdbf48a805d28d37fc : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c96f4cadb6a94eaf94e3ce2c6faa3cda : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784e97cf08464031b5ce1602d9545ca9 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76aaf7b75d8144d89103b7aaf344a85b : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24d51e3265024291b47da900f933a7b8 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80a4c4ad05a24ac1bb7efd10579b2885 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c919e658b36d4b28a27cf3127ce42f60 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a35f40fa99a4560978b5a00e0bf4356 : SUCCESS in 4m 55s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/0fb99b0c58064687844f200f312e3a47 : SUCCESS in 38m 30s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/cc8d9e93525b4c6b96e0bf5d84175ada : SUCCESS in 23m 31s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/c8754ec13b914c3aa6ae177984304409 : SUCCESS in 28m 12s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/4d53d42f3d18495d95937e0b2afcc00d : FAILURE in 34m 09s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/a9b8307830b74358b338c750eb4b8a97 : SUCCESS in 24m 24s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/216026dfcba64d2c9b27395a934169ab : SUCCESS in 1h 00m 01s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8e77405df7c4453a9b46793259f1d88 : SUCCESS in 36m 22s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8a5507f0a14e4be7a55166209a212a09 : SUCCESS in 1h 29m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/436e789be191492aa7c83de6902e0b37 : SUCCESS in 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33d5e92274f7de0fbf792990555ba5591583ec26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-07 07:59:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"159b2dd14962fc4e15e5ba60f1c51012758ee69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 09:32:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d56be8f8e6417bb4f7c8a663824309 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b15132c8cf2843afafe0197faff3f5a7 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21514e2296164d1aae209c93a8c52b87 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/357b7c589bee4bc4b8c9bb5689c96124 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309043323cb04d1fb82333ef2d2dda33 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c9aa846de04bbc83856066f30ee160 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a7548bcf3846dbb7c161a97b92db41 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbc3fece760b4441b2edc72587dec2be : SUCCESS in 5m 32s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/edbd8fab57934977849690e2d86d4ff7 : SUCCESS in 35m 47s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/8a6598f659514bfead30260a7513cf2a : SUCCESS in 23m 22s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/d293fcf7b5504cad8146b4489f46f66a : FAILURE in 24m 46s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/95934e11dcd84e11b89e4c3b6f79e6be : FAILURE in 28m 13s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/df4049f394e348b38e48debfbb426390 : SUCCESS in 24m 39s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/c0f9d778010843ad9257bc1537562031 : FAILURE in 57m 09s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86ed6ca1f34645839fcaa36b222b948a : SUCCESS in 40m 42s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f34f1f07e9b44324953956be443bdfa9 : SUCCESS in 1h 31m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2da765c2e7b497590566c7382fa703c : SUCCESS in 55m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4adc45ea5227101f702345b1ffb0bfe9286d7066","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-22 08:56:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b1af75a8b21e6615692046923558e74a975f12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 10:36:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90f7ada0e9f24a88b9e5f732d892245b : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3696f164068047feb44f50387027490c : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7e6210d23e41a69d7eaf27b0fa237b : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aec97ed7e18c4ffb83cdcbaed14cd10b : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf7dec41b6e440929df508823d854c14 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b78bb4c61917471f8fb0d8248f83aafe : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1034f9ffc65f434cbde973c2091e9b3a : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f5684bf3d947409b615c8baafe59b6 : SUCCESS in 5m 02s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/600f84d220ad41768bd43a6cadbbf69b : SUCCESS in 34m 21s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/0589f375636d4eb2a967fce6e7bf4001 : SUCCESS in 23m 53s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/278e3b7be1d844f0ac36b878bb39f1c7 : SUCCESS in 21m 41s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/f7e059bd604148d7ac84d2760307a824 : FAILURE in 31m 51s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/2e4110758cac465abd3cb894cc957ee2 : SUCCESS in 23m 55s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/502f1dad7b834c58bed7b1eb3fc76a46 : SUCCESS in 1h 02m 59s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc89e854eced43c59d071083ecbde679 : SUCCESS in 38m 17s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1effc42b1d844ecc8e65d144a1189429 : SUCCESS in 1h 39m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0381bee2e6ef44628d408ed175031595 : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f69b13188cdbb981558cbcaac1cdaad764a87ba2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-22 21:41:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5725e454c578f7a0cf784df73c99c8e559aae7e3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-04 08:57:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"86d6b8a3394bf547fd6a94829ed3483021aa13b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 10:38:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3115d816fbac45e88e8f98a3ea2c02b0 : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2fed1514f6a4f99bdc135853a119f55 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4705ca0917694ba280505adefc3c25ce : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc7b1e038d5b43b198ea26b8288743a4 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8bd3f28fde44cc84233d0445578956 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6666a19755a543feac4796bce1bca1df : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f362f6765c3c4078b2399f9d034810c8 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee95b178be3140ef8c017d4c1e977f48 : SUCCESS in 5m 32s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/08b4401242f64d05967f07f5108c2305 : SUCCESS in 46m 37s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/18fb2b338d5746cdbe36a3417e5305c6 : SUCCESS in 26m 32s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/0e3bd4cc3a7d458b97840990b1fff863 : SUCCESS in 30m 37s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/230acfc9827641d0bbf39b2da7bc6ea1 : FAILURE in 31m 24s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/e64958172cef446c96878fdfb13100f7 : SUCCESS in 31m 17s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/e273e2758e574dd693a430ab0140f048 : SUCCESS in 1h 36m 42s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a5d037532bd4f199f314e3d8e00a14d : SUCCESS in 59m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4261893f05554dce8c3e125c86f518a1 : SUCCESS in 1h 36m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/116a69e1986748aebcc59c21d10eef7b : SUCCESS in 1h 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62e3c178e7cecd08911d662ed573cbb9ca67319a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-04 22:13:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6f8daab281c6d756d4b3f6f7742ae9578fed2a29","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-01-10 15:21:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bb58539873c3dabf41dc1620a17c006f65b7a833","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-01-10 15:29:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d28d1d8aae031946f0a5f2960a6dd7f8754efe37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 15:29:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2d748ddb28a742da82e7090cd854ef5d52a6523e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:40:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"df220237ebfe9c5abdc29decab1d2f2f38f02a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:40:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79cab1f4f88a499fa96e5422ae1bae31 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f595671edf9b46318b76da1257b41f39 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef044c5b24d34e579a7b33ea5e438c4d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2990b01d12624136b7b4cf3ec920482e : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a489b388a68140fe8afaf9532a39063f : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d73e12c1935484da206380a90a926e3 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d65177ecc19c443ea121d5e77d3f1ae9 : SUCCESS in 5m 33s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/6ed79bf5a8a6469383b1250d5b34a28c : SUCCESS in 38m 23s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/e00c2f95e74e4624ba5e1cee938115c9 : SUCCESS in 28m 38s\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/7f9ed2f98efa47c2a1506d3dce9f2d36 : SUCCESS in 26m 48s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/6a09773808aa4ccdbb24996a84a3a1a3 : SUCCESS in 1h 05m 28s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1fdc8a655384c57b1a23cabf4458dff : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":5},{"id":"0b2e7e31c3d21c951ea761fd00f3e573fe749207","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:42:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09bd0770c5e430aa55df22699ad30fe : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e172a205a4940f69552142cb907d6a1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"303c829ac33a78596b687aedde6e11e875451aac","revisions":{"721db1065a715879e6a3235468b0d1f8d3d9c17c":{"kind":"REWORK","_number":1,"created":"2019-11-15 09:01:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/694482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/82/694482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/82/694482/1"}}},"commit":{"parents":[{"commit":"7086558f43035b8630acc332005050e86b49bb14","subject":"Merge \"Support list containers by type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7086558f43035b8630acc332005050e86b49bb14"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-15 09:01:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-15 09:01:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I23b0859f81ab32158ef4364228c2bc782d7f6d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/721db1065a715879e6a3235468b0d1f8d3d9c17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/721db1065a715879e6a3235468b0d1f8d3d9c17c"}]},"branch":"refs/heads/master"},"0187e58cf46a83de8b39d2f46f119d64f59e87f1":{"kind":"REWORK","_number":2,"created":"2019-12-04 07:54:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/694482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/82/694482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/82/694482/2"}}},"commit":{"parents":[{"commit":"2178426c0c92348fba42f0db8dfde79a44f5372e","subject":"Merge \"Fix the barbicanclient installation not from source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2178426c0c92348fba42f0db8dfde79a44f5372e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-04 07:54:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-04 07:54:04.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I23b0859f81ab32158ef4364228c2bc782d7f6d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0187e58cf46a83de8b39d2f46f119d64f59e87f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0187e58cf46a83de8b39d2f46f119d64f59e87f1"}]},"branch":"refs/heads/master"},"cfa7c1642d72cde8e90e0ad888d794e8bf560c38":{"kind":"REWORK","_number":3,"created":"2019-12-07 07:59:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/694482/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/82/694482/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/82/694482/3"}}},"commit":{"parents":[{"commit":"d633693bb0848fe73c0be5d7a24bb2883d441d86","subject":"Merge \"Dejokerizes 404 logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d633693bb0848fe73c0be5d7a24bb2883d441d86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-07 07:59:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-07 07:59:38.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I23b0859f81ab32158ef4364228c2bc782d7f6d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/cfa7c1642d72cde8e90e0ad888d794e8bf560c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/cfa7c1642d72cde8e90e0ad888d794e8bf560c38"}]},"branch":"refs/heads/master"},"975ff2ae117427bfc42bedf3d025a8c05657c2ad":{"kind":"REWORK","_number":4,"created":"2019-12-22 08:56:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/694482/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/82/694482/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/82/694482/4"}}},"commit":{"parents":[{"commit":"d633693bb0848fe73c0be5d7a24bb2883d441d86","subject":"Merge \"Dejokerizes 404 logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d633693bb0848fe73c0be5d7a24bb2883d441d86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 08:56:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 08:56:03.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I23b0859f81ab32158ef4364228c2bc782d7f6d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/975ff2ae117427bfc42bedf3d025a8c05657c2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/975ff2ae117427bfc42bedf3d025a8c05657c2ad"}]},"branch":"refs/heads/master"},"303c829ac33a78596b687aedde6e11e875451aac":{"kind":"REWORK","_number":5,"created":"2020-01-04 08:57:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/694482/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/82/694482/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/82/694482/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/82/694482/5"}}},"commit":{"parents":[{"commit":"d633693bb0848fe73c0be5d7a24bb2883d441d86","subject":"Merge \"Dejokerizes 404 logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d633693bb0848fe73c0be5d7a24bb2883d441d86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-04 08:57:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-04 08:57:46.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I23b0859f81ab32158ef4364228c2bc782d7f6d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/303c829ac33a78596b687aedde6e11e875451aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/303c829ac33a78596b687aedde6e11e875451aac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
