)]}'
{"id":"openstack%2Fbarbican~695052","triplet_id":"openstack%2Fbarbican~master~I6aabb1463b001e81e1e6f7d44fb66e9b7ff3a251","project":"openstack/barbican","branch":"master","topic":"drop-py27-support-devstack-default-py3","hashtags":[],"change_id":"I6aabb1463b001e81e1e6f7d44fb66e9b7ff3a251","subject":"Gate on py3","status":"MERGED","created":"2019-11-19 17:57:00.000000000","updated":"2019-12-03 21:48:29.000000000","submitted":"2019-12-03 21:47:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695052-1575409634009-19300d8e","meta_rev_id":"33c5a2a2122ebcfdfe2b8d78c2139fd2559ebf6b","_number":695052,"virtual_id_number":695052,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2019-11-20 20:51:23.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 21:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2019-11-20 20:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-12-03 20:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":1,"date":"2019-11-27 11:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2019-11-20 23:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-11-21 01:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-12-02 15:58:34.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-12-03 20:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 19:50:04.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-20 20:56:25.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-11-20 23:20:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-11-21 01:36:02.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-11-27 11:19:28.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-12-02 15:58:34.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-12-03 20:36:20.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2019-12-03 21:47:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"437b7410224b38368fff3037e1fbe61b5936209d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-19 17:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"905733b5a32383956d63e948ed92aef795d892d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:06:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01dee5dd61b847faa9b28b065ba3bb3f : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40ebe8897598444b93d150deecb07649 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e9b5b65c90415b9515809601f76871 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d85fa375420c43e78dd1fdfa3979d88f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0a379e257f4fe6a19c1ac5abc43b9b : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93e05636dd054b40a41eefcf6329b078 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89df2b7f1c3346c683422329df78cf23 : SUCCESS in 6m 06s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/04b79b5e4df14901a0c989ff615ad8a7 : SUCCESS in 50m 34s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/b72b9c43ff564603928000f0e7b22961 : SUCCESS in 22m 44s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/40dded042d0e440c9894049ea29dbab2 : SUCCESS in 27m 59s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/185cf30ebdcc4af69cbd5a2b62c798f1 : FAILURE in 37m 17s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/27c96bda37ae4239b2714fce5b89a99c : SUCCESS in 25m 22s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/ac433b3305ba454f845990f98a3bf481 : FAILURE in 1h 00m 43s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b7963f3f1444e69233c67a79c70263 : SUCCESS in 49m 00s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d147a0a40adc4e40835e081acf0fbb7f : FAILURE in 41m 19s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9253de5f7aa84ee8a7791ee9c43ebbbb : SUCCESS in 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3859bcf84e2ad593a669aa1c279554f0cf239ce","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-19 19:50:04.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need something like this:\n\nhttps://github.com/openstack/grenade/blob/master/playbooks/legacy/grenade-py3/run.yaml#L35\n\nin here:\n\nhttps://github.com/openstack/barbican/blob/master/playbooks/legacy/grenade-devstack-barbican/run.yaml\n\nLet me get some barbican stat padding in here while I\u0027m at it.","accounts_in_message":[],"_revision_number":1},{"id":"52670fbaa7b054b0286bde9ff12d89e051cd1a75","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-19 19:55:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01ed13033b7bfb0f8ec817fbca74d9e4c41253db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 20:12:25.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"8aa9ceccb358d43d36e725b44db42b4519200f0e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-19 20:42:36.000000000","message":"Patch Set 2:\n\nrecheck dep rebase","accounts_in_message":[],"_revision_number":2},{"id":"b62fda29025baf61d85fdd2cf2f3b8bc46839026","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-19 21:30:58.000000000","message":"Topic changed from gate-py3 to drop-py27-support","accounts_in_message":[],"_revision_number":2},{"id":"44a353171e9d80a6e343c78aad4a63af01a5b0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 22:53:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4c4911173b84653a070e98b54b272c9 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50943aa2ca3c47fdaa481597f4c6a69e : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc6e784fa094aa280ac9be6a4afd7cf : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9916b0e6fe14604ad77b8d3c90e8cd4 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13fa30bcff794ef38f7f10f1bf1b01bb : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c296bce773da4c3985c68c7e9dc27849 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20e272576264e2ea9cf8889bc398bb0 : SUCCESS in 6m 18s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/ca29c0e5a24d40a7a78e9b1180597946 : SUCCESS in 40m 41s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/74228d4abdf4472db1e5019dc5fc9b98 : SUCCESS in 23m 43s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/e0a5db43a84046f6ba9da1265c86e21d : SUCCESS in 25m 18s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/72ef3bb8d3514872a455900a4eb17f65 : FAILURE in 30m 15s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/7b9b1fdeef834fc0a3bd608c0891adc0 : SUCCESS in 28m 01s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/21fad23f7d83431fad3998c5b02a8ef5 : FAILURE in 49m 38s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f0c866409642bfb7887a0a8d9c0c00 : SUCCESS in 37m 44s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/aaf4eb1cee1b48659fd29ff9b94f5b01 : SUCCESS in 2h 02m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de276618389f4a3ea022ece841c6d067 : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0826dcc758fe1792fa694081b6b6f8187ca41dfe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-19 23:05:07.000000000","message":"Topic changed from drop-py27-support to drop-py27-support-devstack-default-py3","accounts_in_message":[],"_revision_number":2},{"id":"55de22dbfbc35407d5cb032810d25c0f5d0cc2b9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-19 23:26:07.000000000","message":"Patch Set 2:\n\nThis broke the same way. The \u0027old\u0027 side is definitely still running py2. I think the dep itself is working [1].\n\n[1] https://zuul.opendev.org/t/openstack/build/21fad23f7d83431fad3998c5b02a8ef5/log/job-output.txt#1765","accounts_in_message":[],"_revision_number":2},{"id":"69b2f988c1f5c115c9562ccea599168c56ef591e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 01:43:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa293166de567e02313306fe8606dbe9d2786ddd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 04:44:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"851083931c5b0e83f899d7f6939cca1e8009ddad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 06:24:21.000000000","message":"Patch Set 2:\n\nthat is because run.yaml is run from stable/train which should be master\n\n- https://zuul.opendev.org/t/openstack/build/21fad23f7d83431fad3998c5b02a8ef5/log/job-output.txt#2111","accounts_in_message":[],"_revision_number":2},{"id":"4720f290140a29efdc5a848049d76642cb97fc0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 06:26:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73cb426d4e3e430c8d158c16f76f26e1 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a81468d6e284f08854a9f441a066f93 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153a3a2e0ebc47a7bb0d21f49951c665 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f7b5f4449ee4e53848349da53f84dc6 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc2a603b0a7e46e78148d1684a6a490f : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cea64cb084e94396bd581a7fdcaa14c8 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2481f7ab6db440599f12e51cdf12b9f6 : SUCCESS in 5m 44s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/e0cd95f9008b48f1ae8617378b6f8a3e : SUCCESS in 44m 25s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/21b88a63b11749a8a4d44ba0da3d452a : SUCCESS in 21m 25s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/d6b9d61753ee4dea8f42ad921a4409ad : SUCCESS in 25m 45s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/0fc33d80a469466c95d4251f65cc6085 : FAILURE in 26m 57s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/8c8be78a7e674f2ea9d1d9ce8929567e : FAILURE in 24m 32s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/a3a17f8e4d054b6d9c417b44077159a1 : FAILURE in 55m 32s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6c15ead734649829b40053ec59b4696 : SUCCESS in 41m 22s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/32f2be2ac77d4caab68f7f9b0fd69017 : SUCCESS in 1h 40m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a120f2df6eea41419bab8d259e1396d4 : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d269ecc9f9644200cf5ce68370a89915f318d079","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 07:50:14.000000000","message":"Patch Set 2:\n\n\u003e that is because run.yaml is run from stable/train which should be\n \u003e master\n \u003e \n \u003e - https://zuul.opendev.org/t/openstack/build/21fad23f7d83431fad3998c5b02a8ef5/log/job-output.txt#2111\n\nnot sure why job inventory is stable train which is causing the issue here- https://zuul.opendev.org/t/openstack/build/d1d06348e4b947cf99066c69498f99b7/log/zuul-info/inventory.yaml#32","accounts_in_message":[],"_revision_number":2},{"id":"70bbffb923591cda6a14f30128580ee33d0c55a5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-20 16:51:50.000000000","message":"Patch Set 2:\n\nOkay, sounds like we recheck once https://review.opendev.org/#/c/689458/ merges.","accounts_in_message":[],"_revision_number":2},{"id":"9c324bdd42830c893acafc55860e8ccfced6987a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 17:53:03.000000000","message":"Patch Set 2:\n\n\u003e Okay, sounds like we recheck once https://review.opendev.org/#/c/689458/\n \u003e merges.\n\nthis is merged, let\u0027s try again","accounts_in_message":[],"_revision_number":2},{"id":"d92f852ed9b79f6f24e5debb97d263fe60319d0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 17:53:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ae63f9c3ab4e018cf607ecf9535d007e63a28349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:51:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67afc2bcb293463d86e093b76fd014c1 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a39ff58c19f647e6824458eec258d57a : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1184e050da444779fa74c21b237ae2e : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/440f6556009746e08e7b58a8ba2fbe52 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c72ee27960441ea8adc7610a3c0e31b : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a7272d5295b41148dd16c5dbc0eebbb : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16da851a80b94e43b1c7cdb38b6bc992 : SUCCESS in 7m 40s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/f665c79bdcbe47688e79f2a7ee7d101f : SUCCESS in 39m 15s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/164a1b84015348d7aed872089f2499fd : SUCCESS in 24m 03s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/6cee9c19d1834b1eb35af01fe6181f15 : SUCCESS in 25m 52s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/c62e6bfcc59041e79b5971014408f798 : FAILURE in 32m 34s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/55c5c31158814dcbb2972f476ee7bebe : SUCCESS in 33m 30s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/9a898296074e4b2b966f3b93f198c910 : SUCCESS in 1h 04m 25s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/508ab0b22b1a4ccdba6360d82e260731 : SUCCESS in 41m 41s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d4a31495d3374244b85dab03d244d9bc : SUCCESS in 1h 43m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5c053df94d4b49689a8c32cc39e13ed2 : SUCCESS in 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ca06789ba64d18f7bfa3a223f14984857f48291","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-20 20:51:23.000000000","message":"Patch Set 2:\n\n\\o/","accounts_in_message":[],"_revision_number":2},{"id":"da57cd9f8d79ad20796f177d9b64c984b6ab4d24","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-11-20 20:56:25.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"a0fefdb30b31be48f2dd6c22701a908acd7d2f9e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 23:20:02.000000000","message":"Patch Set 2: Code-Review+1\n\nit\u0027s all working on py3 now.","accounts_in_message":[],"_revision_number":2},{"id":"2789d3b14b2f7acfab9302f48c81d72c8fe14e82","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-11-21 01:36:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"564e0626fae748be32de2100796d2301e91c8086","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-27 11:19:28.000000000","message":"Patch Set 2: Code-Review+1\n\ntime to merge!","accounts_in_message":[],"_revision_number":2},{"id":"9a655294586dcc8d45c034e76d651332eea82d02","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-12-03 20:36:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"900da5794b319706cc205aa3405db18914acae7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 20:36:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d927b9b0ca3eca3e5bcb45c70f8411d1e783d94d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:47:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/405b3cc7eeba4adbba674768c652cf84 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08515dcdfad446ea639a12b6bb3281d : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4a1d156a05240f1b4b3bc233b02b592 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1232c23fa114b94a6aa0dbf6919c04a : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2db89fcb4304d6e9cc9316897de66cc : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43be71337e44979bf3f0f8454219c23 : SUCCESS in 7m 11s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/f8dafe8f133948ee937cef6147fd190e : SUCCESS in 42m 30s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/b6337b98cd20460e9ae9bd22e9738775 : SUCCESS in 25m 03s\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/9d2ba8d75c7e43a999755567fff2a720 : SUCCESS in 25m 12s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/86ed7e1f882d4ebb9a96172032f5e525 : SUCCESS in 1h 05m 07s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75dfd778ac744c8096e9a0b005b413fd : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f879215a73cab430839986428d09a50a0b4d1f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:47:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"33c5a2a2122ebcfdfe2b8d78c2139fd2559ebf6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:48:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ccecbe779c94cefa3f566be7af39144 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13c32a0db3a88bfbcfd1d76217e6cba8d644f335","revisions":{"56e11b83a926dc4d23179ddf413a9061a3b073d9":{"kind":"REWORK","_number":1,"created":"2019-11-19 17:57:00.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/52/695052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/695052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/695052/1"}}},"commit":{"parents":[{"commit":"7086558f43035b8630acc332005050e86b49bb14","subject":"Merge \"Support list containers by type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7086558f43035b8630acc332005050e86b49bb14"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-11-19 17:53:34.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-11-19 17:56:59.000000000","tz":-360},"subject":"DNM: Test the gate on py3","message":"DNM: Test the gate on py3\n\nA local devstack test with\n\n  USE_PYTHON3\u003dTrue\n  DEVSTACK_GATE_USE_PYTHON3\u003dTrue\n\nand\n\n  PROJECTS\u003d\"openstack/barbican $PROJECTS\"\n  enable_plugin barbican https://opendev.org/openstack/barbican\n\nblew up because uwsgi tried to run under py2.\n\nThe dep sets USE_PYTHON3 to True by default, which may or may not be\nsufficient to expose $problems, but let\u0027s see.\n\nDepends-On: https://review.opendev.org/649097\nChange-Id: I6aabb1463b001e81e1e6f7d44fb66e9b7ff3a251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/56e11b83a926dc4d23179ddf413a9061a3b073d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/56e11b83a926dc4d23179ddf413a9061a3b073d9"}]},"branch":"refs/heads/master"},"13c32a0db3a88bfbcfd1d76217e6cba8d644f335":{"kind":"REWORK","_number":2,"created":"2019-11-19 19:55:47.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/52/695052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/695052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/695052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/695052/2"}}},"commit":{"parents":[{"commit":"7086558f43035b8630acc332005050e86b49bb14","subject":"Merge \"Support list containers by type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7086558f43035b8630acc332005050e86b49bb14"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-11-19 17:53:34.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-11-19 19:55:20.000000000","tz":-360},"subject":"Gate on py3","message":"Gate on py3\n\nA local devstack test with\n\n  USE_PYTHON3\u003dTrue\n  DEVSTACK_GATE_USE_PYTHON3\u003dTrue\n\nand\n\n  PROJECTS\u003d\"openstack/barbican $PROJECTS\"\n  enable_plugin barbican https://opendev.org/openstack/barbican\n\nblew up because uwsgi tried to run under py2.\n\nThe dep sets USE_PYTHON3 to True by default, which may or may not be\nsufficient to expose $problems, but let\u0027s see.\n\n[Later] We think the (legacy) grenade-devstack-barbican job was failing\nbecause it was trying to \"upgrade\" from py2 to py3 and running into\npackaging conflicts. This explicitly sets DEVSTACK_GATE_USE_PYTHON3\u003dTrue\nin that job to resolve the problem.\n\nDepends-On: https://review.opendev.org/649097\nChange-Id: I6aabb1463b001e81e1e6f7d44fb66e9b7ff3a251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/13c32a0db3a88bfbcfd1d76217e6cba8d644f335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/13c32a0db3a88bfbcfd1d76217e6cba8d644f335"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
