)]}'
{"id":"openstack%2Fbarbican~697210","triplet_id":"openstack%2Fbarbican~stable%2Fqueens~Ic1b8c3a4bfd67fff082297b881df66ffb9ca2c50","project":"openstack/barbican","branch":"stable/queens","topic":"branches-stable/queens","hashtags":[],"change_id":"Ic1b8c3a4bfd67fff082297b881df66ffb9ca2c50","subject":"Fix the bug of pep8 and building api-guide","status":"ABANDONED","created":"2019-12-04 00:46:41.000000000","updated":"2022-12-05 18:42:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fcc1181827b901863204aeedf47e3ae66752a6c9","_number":697210,"virtual_id_number":697210,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-12 10:36:21.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"date":"2019-12-13 21:30:20.000000000","_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"date":"2019-12-16 11:03:20.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-04 03:29:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-04 00:46:41.000000000","_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 00:46:41.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-12-04 03:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-13 21:30:20.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-10-12 10:36:21.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"078fcf445f63c7d8a0a67f5bf241e0d3b507b3ec","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-04 00:46:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47186ebda9324265c3dde900acca98fa549be9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:29:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708d880ba6e84a2bae6a44217d6ee263 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78cd55c6e0ce4f2abe761073e1b4d905 : SUCCESS in 6m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1d4c9f33fbe4f64a7ff2612e8f46e4a : SUCCESS in 6m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a07bc729bf1145f4be83c2fade051262 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6902e389211047eea88d9596387cc62c : SUCCESS in 6m 27s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/22c7d13c353e40b5968957e9e515f28e : SUCCESS in 31m 23s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/485a69eeb6f647c8b2b57f702fb1f4f5 : FAILURE in 23m 14s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/758bca1b93d54610807c9db6726c0089 : SUCCESS in 20m 11s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/904c8a154b9c49f79d4715b4f77c66c6 : SUCCESS in 21m 39s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/d73b501a0b8f4c6e8fa91ead3f6cdb91 : FAILURE in 15m 31s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/e023c71278c5464e967c7a519fe664d6 : SUCCESS in 19m 54s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/ef8d5b721c504dadbf2379bba2ddfb05 : FAILURE in 45m 31s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/aae0e38292834808b763c6404708e33b : SUCCESS in 2h 41m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74cff80ef1774b3fb91e445c6db679c0 : FAILURE in 4m 31s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca8bdaee5bba4713bedf1ac2ff1ebed3 : SUCCESS in 57m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c8f5d798d2150e3edab033211721a0528ce73ba","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-12-13 21:30:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d29fcde13b42022fe5031dbdf68a5a32062e1113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 00:16:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dd781c728949b1b99a63041435f431 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59bd1357ba584b8d950df3c7697225a4 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47d53b1ec4aa4ff8adba77ddaa23a427 : SUCCESS in 6m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e54a2a24f65d4071b5a826ddca03bc0c : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3ebb264f7d346cba73f1f7d16100500 : SUCCESS in 6m 05s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/44ca14e005e4428396794054fd3a4f1f : SUCCESS in 38m 14s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/c67dc119ef5d4b21984060333053403a : SUCCESS in 22m 09s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/3968df1fc233454aaba1bcc9bed6120c : SUCCESS in 23m 47s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/d0fe0f413a9e4e9b912a3424e13e28fe : SUCCESS in 20m 28s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/98e11a6512ac40fcaaa22b702ce2524a : FAILURE in 11m 03s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/35ffc066e17c4193ac800f802e08e093 : SUCCESS in 21m 38s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/c45bc857cb804c5b9ba10ec586913726 : FAILURE in 47m 32s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1d227a926a4a47b0aaa28f04c812902f : SUCCESS in 2h 45m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ac63a2e102d43328088b12425504a8e : FAILURE in 5m 51s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93ea43aca3c34c92bea4d6b38371463c : SUCCESS in 57m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24d8337eb577d011be66f9ab81a0cac316433ac2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-16 11:03:20.000000000","message":"Patch Set 1:\n\nIt looks like this change and https://review.opendev.org/#/c/695327/ need to land together. I tried to combine them here:\nhttps://review.opendev.org/#/c/699186/","accounts_in_message":[],"_revision_number":1},{"id":"96b013172c404747cfe9b8c54ddda2e6a1b921ab","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-25 08:57:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a9b8693b67915f90f1dbfa9e80582e9ed0aa16e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:48:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c04600d77043deaa2f3f99a19cd077 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10404fac67114b79ba983cddea0ba355 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5251418feb43446a8aef9f15163d609e : SUCCESS in 6m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f33a1f7d4b8445cca8423e2c76c338aa : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0285492d5854ca5b25ae523b1cfb363 : SUCCESS in 5m 03s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/3049d4fc0a794fd5a66da03a502f28eb : SUCCESS in 41m 15s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/f00c4bfe0cf540648fb9e9726d758622 : FAILURE in 28m 54s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/602d2b8305eb49e38f00cad7d4f1a027 : FAILURE in 27m 07s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/3284a6b046374d14a26c50a02323ce3e : FAILURE in 23m 57s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/6602fe61ec25471dbe9528674f81f228 : RETRY_LIMIT in 1m 58s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/ffb41f82354b4375a78b758f68426cfe : FAILURE in 26m 08s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/1a4b510a5d474bb5b3181d47be451ce9 : FAILURE in 27m 14s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c51f3908b1d6462e8f1fd016cab2ece5 : SUCCESS in 2h 57m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fb35e23b201446e95ec79098a594b28 : FAILURE in 5m 54s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/651cf4ff889d47e69b56ed8d202eeb44 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be20b2312e3cb8f9fb62dbf49000b609a9d4aba4","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-12 10:36:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0416e59e98771422b133a7c00fff725d9d084abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 15:22:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdd7ce97aa2451c89287056549224c2 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d6ffe928d4c49968bc266f0e2edca93 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2432ba8fb981444889fb6daecf2593a2 : SUCCESS in 5m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/066d4f09a2d54b0e9ed2ef290c5229c5 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8abe55ac774a45778addc676dcdb1f3d : SUCCESS in 4m 55s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/7a0d1e764b3b4fb28a3d8e2c4a993dda : SUCCESS in 40m 22s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/4cd3921720704029ae10d81894bab13b : FAILURE in 25m 52s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/58e6ad0831aa4d4eae9e491d61cca956 : FAILURE in 28m 22s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/35957eca5ecc4029ae1aa56b8193e87c : FAILURE in 24m 28s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/45cf5f4b90f14e73b456a6c20bb80be1 : FAILURE in 10m 06s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/6f2ec8b268254e04bad98182a48a2866 : FAILURE in 24m 12s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/164a336b376b4f3db7761ad9032dd853 : FAILURE in 36m 42s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/45778f4346bb4437a568c89990c392d1 : SUCCESS in 2h 57m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b14036eecd154389b0214ce402d98f7e : FAILURE in 5m 25s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20baebbf198544119999c6c1f97bec39 : FAILURE in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcc1181827b901863204aeedf47e3ae66752a6c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:42:09.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6aee62a4c1b269a36fe80f57276d05e25396702b","revisions":{"6aee62a4c1b269a36fe80f57276d05e25396702b":{"kind":"REWORK","_number":1,"created":"2019-12-04 00:46:41.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/10/697210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/10/697210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/10/697210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/10/697210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/10/697210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/10/697210/1"}}},"commit":{"parents":[{"commit":"9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 03:22:40.000000000","tz":480},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-04 00:41:48.000000000","tz":60},"subject":"Fix the bug of pep8 and building api-guide","message":"Fix the bug of pep8 and building api-guide\n\n`sphinx-build` command is not found in test-env.\n\nTo fix pep8, also blacklist the new bandit warning B105, this will\nbe fixed in a followup.\n\n(cherry picked from commit bf95c37b84688c99bda0849230508d955fc62f82)\nChange-Id: Ic1b8c3a4bfd67fff082297b881df66ffb9ca2c50\n(cherry picked from commit 367453b54319b90d2660013a98b15cbd9e0c0846)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6aee62a4c1b269a36fe80f57276d05e25396702b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6aee62a4c1b269a36fe80f57276d05e25396702b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
