)]}'
{"id":"openstack%2Fbarbican~697212","triplet_id":"openstack%2Fbarbican~stable%2Ftrain~Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c","project":"openstack/barbican","branch":"stable/train","hashtags":[],"change_id":"Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c","subject":"Fix the barbicanclient installation not from source","status":"MERGED","created":"2019-12-04 01:07:08.000000000","updated":"2020-02-05 16:46:53.000000000","submitted":"2020-02-05 16:45:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697212-1580921130914-bbef5f2b","meta_rev_id":"9fba149962ff031bcd8382894f8ad6e79343cc79","_number":697212,"virtual_id_number":697212,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"date":"2020-01-31 17:33:05.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 16:45:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-20 14:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-05 15:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2019-12-13 21:24:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2020-01-28 13:37:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-05 15:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2019-12-10 13:48:57.000000000","_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 13:48:57.000000000","updated_by":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2019-12-13 21:24:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-12-20 14:30:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-28 13:37:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-02-05 15:39:10.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-02-05 16:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16883b53375acc5d4c30bfe0ac5c388339cabff5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-04 01:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b99427ee43675e948fedae4526656e6ce8f23f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:59:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/888af25d1f0a45aebe4962c9169247b3 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fe42f0edf344af7aae34efa567ae603 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bf504383a640beb913a42f0a17e74f : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/025c88e30bcd4cd2983e690a1b814d91 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cca4eda48d949d8800781803e3c1d7f : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6391ced9bc740aa84f793e1078a7e21 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7013dc028c546cf82c0ead005fe9ebc : SUCCESS in 4m 13s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/5fdbc6326a174c6a928fbd86e65d7cff : SUCCESS in 39m 49s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/89c96665e08c4d59ba9210455da1b873 : SUCCESS in 24m 20s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/c04fd78f887e46fabc0810850ca56462 : SUCCESS in 22m 44s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/d3b2d6a774df44aca2ff589b9eb112e8 : FAILURE in 35m 12s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/bc1bc8882c2d47a5a885695da1436cb3 : SUCCESS in 21m 56s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/4970a351e7784be882ee0249852174d6 : SUCCESS in 1h 07m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5ecc9851e91740d9a0ff7c541769cc16 : SUCCESS in 1h 47m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55643a56e49c4bab8d1c8cefbde91728 : SUCCESS in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfd138acb5a4f4e359681249b085adc1012286a3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-12-13 21:24:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b88d9680d7af6e2dee3c6a103576fd64955b9294","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-20 14:30:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8762cea4dc28d2f07a512a463effab155a9d58ea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-28 13:37:13.000000000","message":"Patch Set 1: Code-Review+1\n\nDevstack plugin fix. Clean backport from master, where the patch is merged.","accounts_in_message":[],"_revision_number":1},{"id":"f35f734a48921078350b144531511d9ee997ac38","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-31 17:33:05.000000000","message":"Patch Set 1:\n\nWho else could approve this and the other backports?","accounts_in_message":[],"_revision_number":1},{"id":"bf099b5891cba3b7253e293a3b751ab6ddb01dad","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2020-02-05 15:39:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff2c50c10cf70924c0da213b5c6af3501b7d30ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 15:39:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8db90892299d93fc1e6587996ba44b8caba6f21b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:45:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbbcb884c3ca4a329c398d41284ef7c5 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2d5285b2f74801b544825fa8becdd0 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8334577c1e3d4495a2867266047c0a59 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78d63207a72b4f02911614cba3178338 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57b9a81d84c0430e82fbb1088cc7228b : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002392ad21454e1484d6b9238726f8be : SUCCESS in 6m 00s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/5b652be43682423eb73b39a2dee6fd6d : SUCCESS in 41m 30s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/618475d1724947b2aa4297be553ce863 : SUCCESS in 29m 50s\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/a425bb8bf9064a63ae78a87fa1720725 : SUCCESS in 29m 48s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/eb1f9e257d044d82bfd91a09a4498789 : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"36f492392300d52bb3f0a91eeb62e4450d4934bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:45:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9fba149962ff031bcd8382894f8ad6e79343cc79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:46:53.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edae40b024874ad2829b0614fea5a11e : FAILURE in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07be1986b44191600de8b9bf919bef395c3a8613","revisions":{"07be1986b44191600de8b9bf919bef395c3a8613":{"kind":"REWORK","_number":1,"created":"2019-12-04 01:07:08.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/12/697212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/12/697212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/12/697212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/12/697212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/12/697212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/12/697212/1"}}},"commit":{"parents":[{"commit":"9290fa7c46e6930681b97f1eaae53e3fc5391611","subject":"Don\u0027t use branch matching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9290fa7c46e6930681b97f1eaae53e3fc5391611"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-22 16:12:27.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-04 01:06:15.000000000","tz":60},"subject":"Fix the barbicanclient installation not from source","message":"Fix the barbicanclient installation not from source\n\n- follow the standard installation pattern for barbicanclient:\n  only clones if it is installed from source. This way it is\n  possible to install and test barbicanclient from pip\n  by default, additional jobs can simply add it\n  to required-projects;\n- define the repository metadata using the GIT* arrays.\n  They are also defined by stackrc, but they should be probably\n  removed from there;\n- remove the useless call of configure_barbicanclient (the same\n  steps are already performed by install_barbicanclient).\n  Going forward, configure_barbicanclient can be removed\n\nChange-Id: Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c\n(cherry picked from commit 646a0360b97d09146cb3b9a4d9bef927d074c239)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/07be1986b44191600de8b9bf919bef395c3a8613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/07be1986b44191600de8b9bf919bef395c3a8613"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
