)]}'
{"id":"openstack%2Fbarbican~698237","triplet_id":"openstack%2Fbarbican~stable%2Frocky~Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c","project":"openstack/barbican","branch":"stable/rocky","hashtags":[],"change_id":"Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c","subject":"Fix the barbicanclient installation not from source","status":"MERGED","created":"2019-12-10 13:21:40.000000000","updated":"2020-02-10 02:48:28.000000000","submitted":"2020-02-10 02:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698237-1581302908233-2176cf01","meta_rev_id":"f7f14cebe18d8a31545debf5b67936abaad7f23c","_number":698237,"virtual_id_number":698237,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"date":"2020-02-09 22:22:13.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 02:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":2,"date":"2020-02-05 15:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2019-12-13 21:28:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":1,"date":"2020-02-05 15:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2019-12-10 13:49:14.000000000","_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 13:49:14.000000000","updated_by":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2019-12-13 21:28:18.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-02-05 15:39:27.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-02-10 02:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecdbde7ec07eae0ebd28330dcebe986ab2eeda3a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-10 13:21:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"018e34bfc423b7e24789602c71c0db4d57aeb0ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f455a17acd04cb6a6f05bb590be73b2 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1641041bb70e41109140500b88a03b18 : SUCCESS in 5m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/552ebf3722294c6bb44f81f4ac356ebd : SUCCESS in 5m 49s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5680e555e9124823ba1560cdeaf013ea : SUCCESS in 5m 43s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/9593c8122e9740909386f9f925583120 : SUCCESS in 39m 47s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/dcc99fbc9edf49b28d72b920c55f65f0 : SUCCESS in 24m 36s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/9d7afa34c3914223aeecaf6af6f54a9e : SUCCESS in 32m 42s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/ae4445645763456c928d236f01ba65fb : SUCCESS in 22m 07s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/2c0f51d3a3904edb8c34b77c97dfd718 : FAILURE in 14m 10s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/45dd7841e4d041c184d6e510c5575875 : FAILURE in 12m 18s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/d0a607ca0c1241b7b8566d0ff8b72a34 : SUCCESS in 23m 50s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/d746e190078e4fa592f19b5fdcc1d18b : SUCCESS in 1h 15m 10s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/32d353d1ac4a40dabcc5368536623331 : TIMED_OUT in 3h 32m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749a59c33f834c05bac2fafbb9877057 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dad6989722d74f689dc2bda23458368b : FAILURE in 5m 20s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20056f18e36244568c030ad31a424289 : SUCCESS in 1h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"943032323ba9d5ae5e18b42ed17deca0be659a61","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-12-13 21:28:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7bd3d225612963723e51c49363d3fec09b83d235","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2020-02-05 15:39:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"321ab561c057632b9b46768e1e10f80cc4180060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 15:39:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c4ca5369cc1b5e728a39ccafa872899db0fe091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:45:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1fd998ffd3484ba64a69f9cfaeed0a : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16a76dc8e9ba4d89ab64989d1d8d682c : SUCCESS in 6m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e2acbc6b47d466cb263d00fad06781c : SUCCESS in 6m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2201353419984fd194db54a18d74278d : SUCCESS in 7m 30s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/8f905158d5ea4567adfc815e5eb0e536 : FAILURE in 20m 47s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/a2260e6baef043df8a9218a35c669aea : FAILURE in 16m 56s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/edcd5804f883402f9e17a5baba610ece : FAILURE in 15m 17s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/0fc67c0a61924be789217209d03423c4 : FAILURE in 18m 28s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/826db7e6ec704b3cacc45c8b033ef336 : FAILURE in 17m 59s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/2ab5954c929a4118af5562e8da854161 : FAILURE in 20m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30c517b1db974334ac3f3de08586632c : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"23f4a6747d3ad0b450e0efce76427f2eafd63643","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-09 11:32:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95e06d838406dbb40b6e92995b2840d84eb1ad78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 13:16:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51777a996ea44f1aacafd1e101b6bd3a : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6c2949440eb46e19295e2da63bae89f : SUCCESS in 4m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8d0d8cc82024e78803485de9a5a8f3f : SUCCESS in 4m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0647fd770311424398e0bd712805f6db : SUCCESS in 4m 32s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/a1d9debf352e45309bfacb7660d184f1 : SUCCESS in 37m 32s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/ce88ae2102f04155a8c1d21ae0a3e113 : SUCCESS in 22m 55s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/262bed42eb8a4b2984bc1b39b2eaebb9 : SUCCESS in 24m 49s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/73397d421cdb4edbbd936f5a848891cd : SUCCESS in 23m 36s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/b24b39441c3a494eb530581cd1e28195 : FAILURE in 14m 55s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/d9e6e5514bda4df6acf95e35894939a4 : FAILURE in 15m 40s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/675f02913c884be1b6e53d14992f2339 : SUCCESS in 21m 58s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/c2bcc037412e47ee94695e956f85c106 : SUCCESS in 56m 58s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9f2bdd4333d74ce982f133c498ff7611 : FAILURE in 1h 38m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c70e93d6647e45ab90ca947d8f52842f : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a258f5c88aa42738c9cfad6b3967adf : FAILURE in 5m 05s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f1b051c5d5e42e39cb71e233e013ba3 : FAILURE in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"614b3ea8f69f950e37a8ddfb676db9a819cc3c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 13:16:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"faef8f7065e4c90ef1697057e26b6c6d7f8523f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 14:31:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cd71848a4e41399dd8be1d0ad4e96c : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e29e5c45a16b497f91f0307cdaeb26f2 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/651c735d4a994cd796ae11a5a6b69ecf : SUCCESS in 5m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8c7ff1808d364e5eb3c8e8245e4afd8a : SUCCESS in 7m 03s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/4fcbffb0dd43495687225a9957bb6d28 : SUCCESS in 39m 58s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/af76297c67f2494aa16a6a9252c562ab : SUCCESS in 22m 50s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/42b31ba3086a4cf6a79e40e150e729db : FAILURE in 25m 20s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/2d6edac9da7648b397a35eb0e805559e : FAILURE in 15m 59s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/070a9bd5781f46ac9f314bcf5086da1b : SUCCESS in 21m 44s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/9e54a7554df6480d89968a771626cad6 : SUCCESS in 59m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17dc96241d684ec9b5467507a4563170 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a8b74d871ff7763c2c2c3ea6b7ce3f596d14f06d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-09 14:55:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12b2d4ed415745a7d065e1feb121fd199b3296b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 17:47:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434b28ecd81049dfad6143b098967165 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6611dbc85a7547758bf66c8d6d2a0d51 : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a4405532f234cfeb9ba9737724509ed : SUCCESS in 6m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bfe888744e714036b7dd13f28800965a : SUCCESS in 4m 39s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/083b8af713024696aa7ea79f8f14b2a3 : SUCCESS in 36m 59s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/165c8f20f5414d9ab7200c6032afcbc3 : SUCCESS in 23m 35s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/24ae700652e042a3ae201a35f8b16fa4 : SUCCESS in 22m 17s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/3e217e1827c04f3ea8810c1e1a5ce9b8 : SUCCESS in 22m 16s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/ae8b350c39b14d77b6125a98c3c6333a : FAILURE in 10m 12s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/775b2c5abc7740fd9e5c1b4b874682e8 : FAILURE in 15m 04s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/1af9d86937904a2e9e4799f68035f521 : SUCCESS in 23m 33s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/edd85e6ddb3d4d8b9739f11c8ffe39c9 : SUCCESS in 1h 04m 54s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/939336e5b75f43ab975b77013b88210c : SUCCESS in 2h 51m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ce9a53b7dc407b8b82e1de3d32e88c : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a41acbb37e34d089dd0e571e7554985 : FAILURE in 5m 26s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afa0c8801c234364908772f774a68988 : FAILURE in 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60939082d83ccbe8c1eb32e7922cebb83c3d20b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 17:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d6050a15f9092d9692d35ca464b3046b68ddfd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 18:55:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff4314940694e62ab10adc4c1f6ab02 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2dbb8f59e0b24ae9b5ec3ab2c36e054c : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4774f67258a049adaf3a3c996811123d : SUCCESS in 5m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4988888d695b4b0d8dbf021adef576bc : SUCCESS in 5m 28s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/dac22e6b113c4ca1ad9a25d65d5117ce : SUCCESS in 40m 55s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/240bc1d152714d95ace7b4366b0a7cfa : FAILURE in 24m 40s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/b720dbaf43ac48ba9b240b3d2148e790 : SUCCESS in 24m 59s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/bf4a75b7b3bc411ca85b23d87f576ba9 : FAILURE in 15m 52s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/62fe4a09ca9d40f5858a5ceb8ac9ba5e : SUCCESS in 22m 40s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/418adde1b1904d59b48f450e2bd4fdf7 : SUCCESS in 1h 05m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dd4c45a3efc4e9fa29d1b2bcd1e338b : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bfd409c68287f0e40f74346b91fb639946e49c4c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-09 22:22:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20a9fd15abce3aae4468b11db22f87dec22013b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 01:36:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2057e23dfde42eb98f411625700bd41 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85622142cff14eb9b755a7a4e811c3de : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ed278ca0fee44df8979b96f32bb86cb : SUCCESS in 5m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c63a5988fedb45f784efb9a3dd2b26e5 : SUCCESS in 4m 13s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/c3556dcad1fe4cbcab211266337eda3e : SUCCESS in 34m 16s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/1f6b3ff6cf4448c28b536ad321f9044f : SUCCESS in 19m 43s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/79fd4ec0e3b3425f95a4440b6b35630e : SUCCESS in 22m 09s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/d980a8e618c8472ebdb0e4213415e06a : SUCCESS in 23m 03s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/808f6c63d535468992f1894febf36e90 : FAILURE in 10m 21s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/6e65cb13dd304048a5fabde63cd311cf : FAILURE in 15m 48s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/052c22fc2494481c97d59983009835e8 : SUCCESS in 19m 57s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/8a37c01eae5648e393173bcf20eba07f : SUCCESS in 48m 02s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/12e2a379ef0b4dcebdbac280e631af44 : SUCCESS in 3h 13m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bcf16e766974a25a1db464867b38f09 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448c64f8226240839797cf39fcca09c5 : FAILURE in 5m 34s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/496974790eb74e27bcf63bb910db073d : FAILURE in 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab3835cf7421b5dc6bc8de2e2f56b3d8751a59cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 01:36:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf563bf9713aec755bd5f5c76f32954cd1704ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 02:48:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec8610a944c49bbb206103faeb2b5ca : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/466213dd48954590bbe4444fa01ceb7f : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3bad4b61739640768fce56d985233894 : SUCCESS in 5m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae07fe5a515648eea7111fdaacadbcf2 : SUCCESS in 6m 49s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/7a7c5fedb13b454c841a924355eca721 : SUCCESS in 39m 01s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/e9bcd231b835444b9d32edf6841d8b6b : SUCCESS in 18m 43s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/c53ca0d294a7487bac4209247827499c : SUCCESS in 21m 35s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/3936a126510c4d00869293ecf82a8160 : FAILURE in 12m 08s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/fdec7708581743aaae1b71f4cac29ab2 : SUCCESS in 22m 29s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/42bceeab3cd945f6aaa89227f2429998 : SUCCESS in 57m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd462e1259349f89c847de4719610a9 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f7f14cebe18d8a31545debf5b67936abaad7f23c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 02:48:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37bc0d00df8a3944cc78824343a1882ed4028dc1","revisions":{"37bc0d00df8a3944cc78824343a1882ed4028dc1":{"kind":"REWORK","_number":1,"created":"2019-12-10 13:21:40.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/37/698237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/37/698237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/37/698237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/37/698237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/37/698237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/37/698237/1"}}},"commit":{"parents":[{"commit":"6634aea983be47acb2a209cc2c36005e9fda6692","subject":"Don\u0027t use branch matching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6634aea983be47acb2a209cc2c36005e9fda6692"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-22 16:12:27.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-10 13:21:28.000000000","tz":60},"subject":"Fix the barbicanclient installation not from source","message":"Fix the barbicanclient installation not from source\n\n- follow the standard installation pattern for barbicanclient:\n  only clones if it is installed from source. This way it is\n  possible to install and test barbicanclient from pip\n  by default, additional jobs can simply add it\n  to required-projects;\n- define the repository metadata using the GIT* arrays.\n  They are also defined by stackrc, but they should be probably\n  removed from there;\n- remove the useless call of configure_barbicanclient (the same\n  steps are already performed by install_barbicanclient).\n  Going forward, configure_barbicanclient can be removed\n\nChange-Id: Iea1cd3f82c3b38f03f91b0191846e1ddbbfb1d6c\n(cherry picked from commit 646a0360b97d09146cb3b9a4d9bef927d074c239)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
